)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~743540","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Id178a814f0c54dc4d0a9041da0c191fbfa89b4ae","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Id178a814f0c54dc4d0a9041da0c191fbfa89b4ae","subject":"Revert CentOS 8 jobs to voting","status":"MERGED","created":"2020-07-28 13:47:04.000000000","updated":"2020-08-05 10:23:15.000000000","submitted":"2020-07-30 12:11:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743540-1596111087042-b464fa81","meta_rev_id":"228aa0c0b5929c431fe8b0b4dbb8ba7a30d21309","_number":743540,"virtual_id_number":743540,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 12:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-07-30 09:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-29 13:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-07-30 09:49:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 13:52:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-30 09:49:05.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-07-30 12:11:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e8c4666630f6a18e4eaa5a5a8f4b89ea7a6b4f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-28 13:47:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c76713a40df6836ad56e8b61ee4a7bdca2f17b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:25:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fbcbbf2bcf16450999644818077aab08 : SUCCESS in 7m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8e9331693df44ec08d131247c2e3d5a1 : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df1e09e615e14c7ca0c94019b3fe13c7 : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b752f17afb274afea16343ff3d675ab4 : SUCCESS in 1h 06m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf4c4a596a324a7faf01b9647604f8d7 : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a06796cd6936406cb72815aeeae8c66e : SUCCESS in 1h 17m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/abaeb2307f0747a992c1eb3010f06407 : FAILURE in 2h 13m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11005a5d52b04a168c7b622f08b035e6 : SUCCESS in 1h 41m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5c1b766bba664d17909aee7e86218b25 : SUCCESS in 1h 48m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/793f8898058249419dfba68586298257 : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb48b46385b242d8993a2fbec26cd784 : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14fc090e523e4cf899eaf4b2b84baddf : FAILURE in 53m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e7787a6b46d94bad8e9824f6f55b6106 : SUCCESS in 2h 03m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee0a9ba7969a405bb70562b93d3a4451 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/967ebeabbe744414968c0ebe1a551887 : SUCCESS in 1h 41m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1540b54c4cdb4bfca1dfa96b27d99ae1 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6daf64ba4884dc467aa1d0c2d08fe0452f42d26b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 10:30:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aa4e008e54b0fb648b99c7f017ff7816aecbe29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 13:40:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27a621c0d0264167ad6ef4113388db3a : SUCCESS in 11m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5c015daae38841f09bf899906908180c : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/18da94dc796b473ebd210cf4fb9c05e0 : SUCCESS in 1h 56m 10s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/45fadab8279a40d9843b341a044f4d7a : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84746d2010c6450fbb1782a86655e742 : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e967df7dfef43dcaa84dc706d1e48ce : SUCCESS in 1h 17m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f211137455d4d0aa75ee545b8c936a8 : TIMED_OUT in 3h 04m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41e1fea10cdf4d519f6b501635324916 : SUCCESS in 2h 08m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/51958943a3e145548c86bdbf1d4d95ae : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5dd7e486cb7b4381b491356d3dbd764a : SUCCESS in 1h 12m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e3adfe393e0455fb0169c2823ea9162 : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/253ba846c32a4ab484150efdafb66e26 : SUCCESS in 1h 04m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1deee5b6c9244401a0f270695d7844fd : SUCCESS in 1h 39m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9e567b5d36904d749e606f8ddaab9802 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15771b7a72d340d08f33bab260adf022 : SUCCESS in 2h 06m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd06d5c284943aeb05ae87c76050c06 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"63bd29ce6cd17d5326d463f7d4ee138caf0bea04","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-29 13:52:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e85213d17f9e2cfc7da134bc2f7130e67ddcbad","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-30 09:49:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7093d17018f4e859e7a0e0dd6aa0d03bd5e59f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 09:49:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09177900e65f309ef5bd99b6347096d063d2600a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 12:11:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d073065a766d4ab8a7a0817e24a5219a : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/591318d02b9b41d796170c754688c784 : SUCCESS in 1h 28m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dcd0b9bd242c44508ba318aaf353d5ae : SUCCESS in 1h 44m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e38be086f6bb49af9ac886f36ad1ea14 : SUCCESS in 1h 48m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/566275002acd47d08e5ce8225bef362f : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dca795c61cbb4b729267a1c2ad4458e5 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/66b21d81bb3548b0aae17eda4d8dd4b8 : SUCCESS in 1h 17m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/993aec782ed04abf9f8fe51f53564e90 : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea450de6a4954cd4be47989942853ed8 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08218ea51c274e5082db60827dc07e32 : SUCCESS in 1h 02m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-7 https://zuul.opendev.org/t/openstack/build/822cf6b0d95a4a0d900fafa82fbaac5e : SUCCESS in 1h 57m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d42ac3a317b34294afa6cf84d343252d : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/956656d8f6d7438aa179aa052ad0cf22 : SUCCESS in 1h 27m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605d4dad4e394b7496ec708d7ec3879b : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":2},{"id":"048c42474924b45e3d56ec4d11c0da62b49d806c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 12:11:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bf69fbca484d31f0b6f744da8c711538cfdd22b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 12:15:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197f4a9b33bd49e28bb85d725e4a0e21 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4c7b7cb184a08384cc047ce121b32edebe4d100d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-05 10:23:15.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit e322ddd4c283d46ca60558eaf8f825510beed12a","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"70e6c6d3ecdac4484d9212837c7424e0db36e826","revisions":{"1a9b8abc0fede6dfd08dc61c99db50255aa34577":{"kind":"REWORK","_number":1,"created":"2020-07-28 13:47:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/743540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/40/743540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/1"}}},"commit":{"parents":[{"commit":"daf883d3170cfb49443a2978fcf78823e9ce227a","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/daf883d3170cfb49443a2978fcf78823e9ce227a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:46:53.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:46:53.000000000","tz":180},"subject":"Revert CentOS 8 jobs to voting","message":"Revert CentOS 8 jobs to voting\n\nChange-Id: Id178a814f0c54dc4d0a9041da0c191fbfa89b4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/1a9b8abc0fede6dfd08dc61c99db50255aa34577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/1a9b8abc0fede6dfd08dc61c99db50255aa34577"}]},"branch":"refs/heads/master"},"70e6c6d3ecdac4484d9212837c7424e0db36e826":{"kind":"REWORK","_number":2,"created":"2020-07-29 10:30:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/40/743540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/40/743540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/40/743540/2"}}},"commit":{"parents":[{"commit":"daf883d3170cfb49443a2978fcf78823e9ce227a","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/daf883d3170cfb49443a2978fcf78823e9ce227a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:46:53.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-29 10:29:45.000000000","tz":180},"subject":"Revert CentOS 8 jobs to voting","message":"Revert CentOS 8 jobs to voting\n\nThis also fix ubuntu focal distro installation\n\nChange-Id: Id178a814f0c54dc4d0a9041da0c191fbfa89b4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/70e6c6d3ecdac4484d9212837c7424e0db36e826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/70e6c6d3ecdac4484d9212837c7424e0db36e826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
