)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~759883","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Iab8ed570cb6f60afc09f4ae47b97c49c8e61072d","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"api_threads","hashtags":[],"change_id":"Iab8ed570cb6f60afc09f4ae47b97c49c8e61072d","subject":"Reduce number of processes on small systems","status":"MERGED","created":"2020-10-27 14:55:54.000000000","updated":"2020-10-30 01:41:39.000000000","submitted":"2020-10-30 01:40:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759883-1604022017007-8dccae41","meta_rev_id":"3bb42c82c5a5946765b9105a6fd3e77045c40340","_number":759883,"virtual_id_number":759883,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-10-29 19:32:16.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 01:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-29 08:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-28 17:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-29 08:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 17:48:52.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-29 08:53:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-30 01:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1408e06f72b25e8bd22377f153781dbfb6f1cc22","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-27 14:55:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2be196bb439b2a0e74c8de6596c6c3136f834bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 18:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/10f4660140e94d77b3a2b200d26cb067 : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ef539363d29740c582c9e7dca7ff7e8a : SUCCESS in 1h 47m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0efaba1b7de74e42942e5317b6549013 : SUCCESS in 2h 25m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e80685d2e5504ed084e902f02e8761db : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba66d955e7d545298e0902299d70ea72 : SUCCESS in 1h 20m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/35913e27f8fb490bb5548fc25bd247cc : SUCCESS in 2h 46m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3461f2721c4d445b92011c5c2f1112da : SUCCESS in 2h 03m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d10dad2e6414dc4b9f9ad3ca2cdab92 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/598b8909bee44c97aeacf7f7a0e82b07 : SUCCESS in 2h 09m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0cd1c77919b847918545f805d2f1e4b4 : SUCCESS in 2h 26m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bbbeb22a34324761af8b65a48a3e1187 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea560721cb24e8499d2afb10cb0df05 : SUCCESS in 2h 26m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b6f52a99f143ab9d420ace1103965a : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f5380b776cb679eecace4bdcd2fa1c473579586e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-28 17:48:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb9cab00f69ba8829c70fdddb71e377add8e401e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-29 08:53:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f13cd0e04e58248ff4d3f2b99ac1ba2273bf976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 08:53:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4bcb8fe6ddd50c2b5708911281a37c7615a90ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 11:50:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84b02c3822b34f95891d4be41a63646f : FAILURE in 7m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dffe72043e2f4a03bfd7456d53e02a03 : SUCCESS in 1h 41m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e9bf6d7b78c24adbbdb5a68b85c0955d : RETRY_LIMIT in 8m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/763e24b07a2f4127aea811be4e100dcd : SUCCESS in 1h 46m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd8f65fa4fae4ff8b1823a0388f3b503 : SUCCESS in 2h 08m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cf9fdc4bb8354f9d8122209fa0fc2eec : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a38fd0321e84f1e9571343d923b8a22 : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f826f0e729d64d71a63e2e5271f71d5c : SUCCESS in 1h 44m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9fd84ebebe994451ae0ab1a3c9555d8b : RETRY_LIMIT in 10m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8be660fb6e924634b552d87667e988e3 : SUCCESS in 1h 30m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f556394d63494ca30b164aa385b556 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cb6c249766fdb10a419ba484af84700821bf9c58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-29 13:13:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c068b6cd364cd5bfe52c458dd34feec96427ee36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 16:18:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a79e08c4cec34eb29797e4c35f6a1ba4 : SUCCESS in 9m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6a8b9d3ad5444306b8d68bc06cfc6b74 : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c6e3b2ca5d1243f19ae3e5b77a0e94f2 : SUCCESS in 1h 45m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a008aff48a8447ba823b9d0e1e60ef0f : SUCCESS in 1h 37m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/810231b08ba242a9a13fc060122c26da : SUCCESS in 1h 25m 16s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e975ffba9f4432586ae2c5ae31cb8cd : SUCCESS in 2h 46m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb96db996409467293d891ae0a650dcb : SUCCESS in 2h 10m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/940a959133314179b304f5730df73b99 : SUCCESS in 1h 21m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b929ae6165542a8a7c972e5799accdd : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/160ba2f59f2d4752b2c895cb74970791 : FAILURE in 2h 54m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ed84707f811048acaf39e03634b4e5e5 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be0be5438d9f4e8f8229cd044d300baa : SUCCESS in 1h 33m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81639d909974678b851506517ef5e94 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"729cf1963936f8b32b46670d8fd5240d947b4536","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-29 19:32:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de44bbeded4075b009b860e65631cf2bc6c5b56b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 22:35:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7fad5b037635464f98599049ce1b247d : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/42e581d64d9444aab2c43d151121cfae : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8eb962c7eb00450db25527b447249c4f : SUCCESS in 1h 44m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11df421caf904b8bbc97a09ae0838228 : SUCCESS in 1h 37m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65fd6a8c488c4996a798d6d2b05c61a5 : SUCCESS in 1h 21m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e53b1022dad94944b6306605374855cb : SUCCESS in 2h 39m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/964dba99589e4762bacbb3e5ed17b42d : SUCCESS in 2h 04m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f627b54d98a04300ab9bf767f07ad615 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a73522b2cbbb4d62870413aab4ac60a9 : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/132174f0d92f4604903e4f638b65e1a4 : SUCCESS in 1h 48m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/00868d504a544b0fb95b578377a49dce : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/212edb033f6648a1920b2e0c6b4d4740 : SUCCESS in 1h 23m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a099e189064b28b9e539f4ad584916 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a697bb9982170d0e536ff18a55a1386734ce0741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 22:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b5c1a1866aee452009fb8eaf83cad9afe6aae8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 01:40:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a15f2da33e01429882c2c3b6dc9f980c : SUCCESS in 14m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/215e9b0baa4341e6928c4b20a21e2134 : SUCCESS in 2h 10m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8f5eb1b1beb24f4ebbd6822899cda3e0 : SUCCESS in 1h 57m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aad520d0ef0b49308ca5cc86b32db49e : SUCCESS in 1h 42m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a91803fefca744d6a08952829f034204 : SUCCESS in 2h 51m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0e8f42717fbb4c49bf1f547206c5402d : SUCCESS in 1h 23m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5749a6218c1475f84ec247fa44bc3c0 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5368a9071271466bba5fbb6383989993 : SUCCESS in 1h 56m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/004c965b1e924d039c961a309a318787 : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d32e0beb3f746b7b5df0dc4f4b8cfd3 : SUCCESS in 1h 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c582eea8e0954ba795883260a525d13b : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"73c89a166c2fdf3bc9e5787b24d552a7a84b8650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 01:40:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cab066fedec6b87782c2834498e5d405c9c67766","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 01:41:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab35bfd47dff4db18bc771c221e788f4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbe63486b2287803cdc81fff77111124a38f49fb","revisions":{"cbe63486b2287803cdc81fff77111124a38f49fb":{"kind":"REWORK","_number":1,"created":"2020-10-27 14:55:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/759883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/83/759883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/83/759883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/83/759883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/83/759883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/83/759883/1"}}},"commit":{"parents":[{"commit":"fdff574188a4a62b43af070de9bca54ec027f9a0","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fdff574188a4a62b43af070de9bca54ec027f9a0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-27 14:55:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-10-27 14:55:38.000000000","tz":120},"subject":"Reduce number of processes on small systems","message":"Reduce number of processes on small systems\n\nEven the most modest 4C/8T system would run with the maximum 16 processes\ndue to the calculation being VCPU*2.\n\nWe devide amount of CPUs to number of threads for hyperthreaded CPUs\n\nChange-Id: Iab8ed570cb6f60afc09f4ae47b97c49c8e61072d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cbe63486b2287803cdc81fff77111124a38f49fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cbe63486b2287803cdc81fff77111124a38f49fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
