)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~761922","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I6d7b5404eb5b706683b1b15575cc880a2cd45571","project":"openstack/openstack-ansible-os_aodh","branch":"master","hashtags":[],"change_id":"I6d7b5404eb5b706683b1b15575cc880a2cd45571","subject":"Remove duplicated aodh_db_setup_host","status":"MERGED","created":"2020-11-09 13:46:10.000000000","updated":"2020-11-17 12:40:16.000000000","submitted":"2020-11-17 12:38:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761922-1605616693848-b2dd4156","meta_rev_id":"d88d2546b42c5a194f5f21403921bb8cab93a3e4","_number":761922,"virtual_id_number":761922,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-11-17 07:38:48.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 12:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-11-16 09:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-10 22:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-11-16 09:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 22:34:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-16 09:20:55.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-11-17 12:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eae19d2b318caeb2d36883fa4198cf5e6fe6f8dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-09 13:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"946e202176979fe0d3ee24353f905b6be7bd0169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 16:40:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc8d37c165234333b0c7b2a0879c4cdb : SUCCESS in 7m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3f556e81f7a944389189a6d04ce658eb : FAILURE in 1h 25m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c657a08611e147a999cc127c1be4faf9 : SUCCESS in 1h 51m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f36825fd8e41618403843eda431e67 : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c416a575ef4490eb4e2412f12b4d89b : FAILURE in 1h 46m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d726c8c0cfe2462ca683b8f7d79706d9 : SUCCESS in 2h 27m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d394f745be3d4e3f825b7c37bd0a29d8 : SUCCESS in 1h 59m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c2f2079794664049bcb1d73d8fcca8d9 : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13ecbd2477b44e7bac7bf8cc81daad67 : SUCCESS in 1h 14m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6c5735d10e5d47759db4c73c4011b8f8 : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a35b926b04844bccb13676136ae878b6 : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9045494c5850445ca399b7a54da03b99 : SUCCESS in 1h 18m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9570a2b5354647bbebbd196c426962 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"de97088a1c390794861aa8eaa5c1e45a4b4d6a1f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 12:09:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6ca3ee860dc1324188925f9289fe79f1df2a6afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 15:27:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cbf772a9c88746929b543892db1014e3 : SUCCESS in 15m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/149228ffbe3b44dbab633bd5032efac3 : SUCCESS in 1h 28m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f88eff59f4614df282e24e34aaa3df49 : SUCCESS in 1h 44m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea1bf392a75140b6ab74063b09424997 : SUCCESS in 1h 36m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f6491fcbb043888728af0a83d66137 : SUCCESS in 1h 24m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5b3e3ea67add435c92b69787eeda6515 : TIMED_OUT in 3h 06m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb859166b0084caaa741f7f6a91e220e : SUCCESS in 1h 58m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9a9e4ecfeb2749baa4f5142951c05cde : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfac14ec6621465694c295b444b756e1 : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5873ef4abb4349a6b53251ae74fd3887 : SUCCESS in 1h 48m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07a015ffe5a245da87f7d8e2368acc6c : SUCCESS in 1h 39m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/287c2a37043f4942a2a7f272304c0895 : SUCCESS in 1h 27m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b0d457e2a24a539ffaf80bb5af54f3 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"941fe76d2b07d57151947b4ef03aa2fde9360d2a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-10 22:34:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62a15cc8a5d6a8b4a40a54ed26e3735b49143db4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-11-16 09:20:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9fa13f6d68c0410f620670080fabde9b7aea8ccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 09:21:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3d9694e87e93eb9835e7d850d8deb6bb4e856fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 11:44:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bd96e6cf5ded429ab9cdeb4cb2b7b077 : SUCCESS in 10m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f9146af9f0aa49678a632ac75bbf5737 : FAILURE in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e7af113d2cc34e4ba19880de1d9d62e4 : SUCCESS in 1h 55m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f8a7994d9664dacb28f443c9ea61d90 : SUCCESS in 2h 10m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d699e3801b6c4088a600e97fbd2e4344 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bc266529fcf841a98baedc4e5785ce2e : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cb19c44b80e4dafb6c02d68b2d2cba4 : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/57773d6de7be4da097b67b31359bfe0e : SUCCESS in 1h 57m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7a9fc206222a42f2a2ac7ca649ad32e9 : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38c5377ee3d74460aa5983d46cf75645 : SUCCESS in 1h 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304b107a0c794dd38e78db0020214678 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ed90a19872d0f07beb70814a257a8bc1da06b2e4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-16 14:11:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dd0379585481df01ae512f884aa8dfd3d4974041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 17:04:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1daf48378b5e4c4581fd8cd4493c420d : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cfb6b22fbde3406d8201e5899276748b : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f319e81cf48848c89940c0c2e56b6bcb : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2533f92d0c034bff8cf3ce61ceca6d76 : SUCCESS in 1h 35m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40582b20d33c4fad9f78f441b5c8494e : SUCCESS in 1h 30m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f26731b0d2fa4ce2930a26e93807a63b : SUCCESS in 2h 38m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aea5b628d9af474198343d25fb18aa46 : SUCCESS in 2h 01m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3d87205da4e141678977e4501b7013c5 : FAILURE in 34m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/651aaf2055ea40278f30900e85cdcb99 : SUCCESS in 1h 18m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3622df1f567448d5a9baa0e0999143cf : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/97df6b11a6cb458392e852a96fd53df5 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/044716a1428941868ddb82628e18512a : SUCCESS in 1h 29m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d652b2d04164891b6af193563d6fde4 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"aa171de1d8462c7314f0fb7ce32322a7d208143a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-17 07:38:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e10391ec22458ec5a81a2c25bc4a60ce6c13f2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22a9eb2446a94b1e9eb1c25b65fa9591 : SUCCESS in 11m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b70407b96ff47239a24f729e50bfafe : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/27543a1aafa3442891fe37469a25cd33 : SUCCESS in 2h 14m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b8abeccadde4223b5762b37c46a0446 : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cc7dff28caa494c8a1cb83d2fa1d9a6 : SUCCESS in 1h 27m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/abff1ffa9c264d27883c83479c798996 : SUCCESS in 2h 48m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6b82f365e2c4b8d8d54475cf674c13a : SUCCESS in 2h 40m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3a9f5c1ca0ff422cbff18655dff4beb9 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35f9a21bf3a44278e9afa0934a26143 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4a22b98276064741a5fc2145036a3f68 : SUCCESS in 1h 48m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8762a6cfb8e441899a43974d0af9dae4 : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bfbc2ebb78a4581b2c7fbcd6527b5b9 : SUCCESS in 1h 18m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c498e0cf79d64844b3341dc06e50a3ad : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a18d54d4c3aa5ac9408ef5062171bcdef84bd816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 10:38:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d6f8decd96540a177f245ac23f5134271b89ea8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:38:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9ebbab73328b2c9e134d4f99e0f96dd1ec505a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:38:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/caaaeeb2dc794cffb30d58d48ad7aca8 : SUCCESS in 8m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9929bbe2674a48688eb6bb47450ad490 : SUCCESS in 1h 31m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/816671eda21c4aad83af00dc32de4741 : SUCCESS in 1h 52m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f89100c0e8fa49bd975509fc1817e6f3 : SUCCESS in 1h 39m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc2163e14ced4ffba553b2255940893c : SUCCESS in 1h 24m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/745b3d44d5fc460ea4db2d4fe34a262e : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/738cb5414ec54a959438746f965bb167 : SUCCESS in 1h 09m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4543cedb1216492aa049167fcee9327d : SUCCESS in 1h 46m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ec784875ebd6462fad6e9194178c89e9 : SUCCESS in 1h 24m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cf6f257f2fc4c8181982a9f9e7136f3 : SUCCESS in 1h 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f75d22c905845d0998861673099060b : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d88d2546b42c5a194f5f21403921bb8cab93a3e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 12:40:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c47241eaa447c8ae79f01d356f87a8 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac880a280292913f26eba373a2f66400993335d2","revisions":{"29afb2d16a439ee42421b2692fc30dad1a5dab4b":{"kind":"REWORK","_number":1,"created":"2020-11-09 13:46:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/761922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/22/761922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/1"}}},"commit":{"parents":[{"commit":"fdff574188a4a62b43af070de9bca54ec027f9a0","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fdff574188a4a62b43af070de9bca54ec027f9a0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-09 13:45:59.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-09 13:45:59.000000000","tz":120},"subject":"Remove dublicated aodh_db_setup_host","message":"Remove dublicated aodh_db_setup_host\n\nChange-Id: I6d7b5404eb5b706683b1b15575cc880a2cd45571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/29afb2d16a439ee42421b2692fc30dad1a5dab4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/29afb2d16a439ee42421b2692fc30dad1a5dab4b"}]},"branch":"refs/heads/master"},"ac880a280292913f26eba373a2f66400993335d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-10 12:09:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/761922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/22/761922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/22/761922/2"}}},"commit":{"parents":[{"commit":"fdff574188a4a62b43af070de9bca54ec027f9a0","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/fdff574188a4a62b43af070de9bca54ec027f9a0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2020-11-09 13:45:59.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-11-10 12:09:06.000000000","tz":0},"subject":"Remove duplicated aodh_db_setup_host","message":"Remove duplicated aodh_db_setup_host\n\nChange-Id: I6d7b5404eb5b706683b1b15575cc880a2cd45571\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ac880a280292913f26eba373a2f66400993335d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ac880a280292913f26eba373a2f66400993335d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
