)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~764009","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I7aa49677375e3b91abfbebd686efa718a6f4ae8b","project":"openstack/openstack-ansible-os_aodh","branch":"master","hashtags":[],"change_id":"I7aa49677375e3b91abfbebd686efa718a6f4ae8b","subject":"Dep\u0027s should be restricted by tox-constraints","status":"ABANDONED","created":"2020-11-24 16:11:48.000000000","updated":"2020-11-27 01:07:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe81bc198b72e7a3c93c1c3566da7076d77b7109","_number":764009,"virtual_id_number":764009,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 16:49:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":-1,"date":"2020-11-26 13:44:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 16:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 13:44:20.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"}],"messages":[{"id":"7aa6cbc09c2c08a5b18fa843e86f29c53e4a9f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-24 16:11:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cc41c92f116eaa335fddd93611e900cd00cbde8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:49:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1a6160bbf4140dd956f03c4824e51fd : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d112d021093e48c2b52f42b68110e19f : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ab99ba6811464bb42a20e82923be94 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"05987a5fef1612eee93ac874ff233a1a42b09805","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-26 13:44:20.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"fe81bc198b72e7a3c93c1c3566da7076d77b7109","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-27 01:07:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e62d24871b76fdee95e221ef218d92102e5f51d5","revisions":{"e62d24871b76fdee95e221ef218d92102e5f51d5":{"kind":"REWORK","_number":1,"created":"2020-11-24 16:11:48.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/09/764009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/09/764009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/09/764009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/09/764009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/09/764009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/09/764009/1"}}},"commit":{"parents":[{"commit":"13ef8fa191c283c4f85579b1b7a5f884594119af","subject":"Merge \"Remove duplicated aodh_db_setup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/13ef8fa191c283c4f85579b1b7a5f884594119af"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 14:53:12.000000000","tz":0},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 16:10:06.000000000","tz":0},"subject":"Dep\u0027s should be restricted by tox-constraints","message":"Dep\u0027s should be restricted by tox-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s tox-constraints.txt\n\nChange-Id: I7aa49677375e3b91abfbebd686efa718a6f4ae8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e62d24871b76fdee95e221ef218d92102e5f51d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/e62d24871b76fdee95e221ef218d92102e5f51d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
