)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~765934","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I7b316fb0508d021f6f2af2c074175fc0bd87e9c2","project":"openstack/openstack-ansible-os_aodh","branch":"master","hashtags":[],"change_id":"I7b316fb0508d021f6f2af2c074175fc0bd87e9c2","subject":"Remove centos-7 conditional packages","status":"ABANDONED","created":"2020-12-08 09:50:56.000000000","updated":"2022-01-04 13:17:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e2b0baeaa86952d848f6ea59d36a19665e4715ec","_number":765934,"virtual_id_number":765934,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-08 12:04:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 12:04:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5f3cadd94ab30864b6874405856ca63ebbf7c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-08 09:50:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec9075afef7eb781f5ba4ad01acb442ab94b917f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 12:04:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd78df0a4a9c4ac4a7740366a19ffcfd : FAILURE in 9m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2bcdd7c4ef494228b2d24fd19cf7faaa : SUCCESS in 2h 00m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ac3f4910f9624b47a33ac107bd6594a9 : FAILURE in 43m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0449672445a43f8a833fcaa977e66a3 : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06ee6ce516bb4b7fafc12fdae1f0c810 : SUCCESS in 1h 35m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cb9ee76a617e4150ac7c92e696e7a3b7 : FAILURE in 40m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e71d78e11ac4161a6e09d0d06f81d59 : SUCCESS in 2h 06m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/baed72b0c1fb4a2283de3cb47d711ff7 : FAILURE in 59m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c1388449254113b2c21926c1ac375c : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d565cc6714a8474bac36348c9192f363 : FAILURE in 34m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a09f50fcc07940c59a2b15770da8386d : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b2016b6de0840acbb17a88ae988d7f7 : SUCCESS in 1h 27m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aea55078b0d4c1da6db0f55832ceef7 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"231c6cca2a69fa50c885ccc128540043fb5ac9db","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-12 21:09:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6bccee7e4313ce9d48d5b32bf63a1261777ffe85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 23:03:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1af5640682584f08a2e3eb20cc1edfa5 : SUCCESS in 8m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5fc1adcf06ab4fed91dc27472d1640aa : SUCCESS in 1h 29m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/095b9e31caca48e8b4d64f75e0d98240 : FAILURE in 1h 29m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/baba3c5de53b41c7998326371d8b2180 : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e157238b5d4b4aa594245e12093d2d6a : FAILURE in 1h 11m 09s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/aa6260cf62b94aae8a1cd131fee293fc : FAILURE in 44m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9fc0a8db8314d6c965562810c6afe4f : SUCCESS in 1h 47m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8fbe80b89dab47a2b7dc7fea7a4cd712 : FAILURE in 1h 06m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3987e12bc53a40118ad4841ad6a1a4ff : SUCCESS in 1h 02m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1f2c8c879ca84cd9953f4b65612b080d : SUCCESS in 1h 45m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b3a440ba22844a78ab6ebd833bc9fcfe : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4e5648a9e143a3bb657e1ea40de115 : SUCCESS in 1h 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd73ba221a014728a7f9ca4badb19356 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"19719b4a0a7c39a266f028dac2f1dd1aedac9749","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-12-13 13:37:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83cfd74ce110145a3a636bed95d62f806af68202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 15:52:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a139667a27794c11865cd0215ea1cd48 : SUCCESS in 8m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/48ad88e45f6f4cf6826a8d9ba54c999e : SUCCESS in 1h 33m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/25a83836b4a548ca83d090058d54f0d2 : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61dbb0c92f074e3497154c37d037a531 : SUCCESS in 1h 38m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2d94c42a2da408180f11db32e7ca9f5 : SUCCESS in 1h 27m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4f4f37b4fec34f2ea86079003b263fc5 : FAILURE in 35m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2663364d58644199c493d416b7fc1ad : SUCCESS in 2h 08m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/08c9ebc424ac48389babd6636243c2e1 : FAILURE in 50m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0031703b42ec4b598714cb75e538a0a3 : SUCCESS in 1h 08m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a6d8c001042040a899f6942dbb397635 : SUCCESS in 1h 45m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6024508e11034637bd6ef51101bcc207 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc8d5c5b08d44dcfa8c00fa75ed21044 : SUCCESS in 1h 03m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41b9acaf2804597b9ec4146d328150f : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e2b0baeaa86952d848f6ea59d36a19665e4715ec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-04 13:17:53.000000000","message":"Abandoned\n\nalready merged with another patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea0b2578801c1e8f996c050a40a93f3576ffbf17","revisions":{"ea0b2578801c1e8f996c050a40a93f3576ffbf17":{"kind":"REWORK","_number":1,"created":"2020-12-08 09:50:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/34/765934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/34/765934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/34/765934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/34/765934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/34/765934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/34/765934/1"}}},"commit":{"parents":[{"commit":"13ef8fa191c283c4f85579b1b7a5f884594119af","subject":"Merge \"Remove duplicated aodh_db_setup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/13ef8fa191c283c4f85579b1b7a5f884594119af"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-12-08 09:50:32.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-12-08 09:50:32.000000000","tz":0},"subject":"Remove centos-7 conditional packages","message":"Remove centos-7 conditional packages\n\nWe don\u0027t support centos-7 for Victoria so remove extra templating\nin package variables.\n\nChange-Id: I7b316fb0508d021f6f2af2c074175fc0bd87e9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ea0b2578801c1e8f996c050a40a93f3576ffbf17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ea0b2578801c1e8f996c050a40a93f3576ffbf17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
