)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~772259","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I73773536938dff9f768381a333c5fa02c98e9d55","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa-new-pip","hashtags":[],"change_id":"I73773536938dff9f768381a333c5fa02c98e9d55","subject":"Move aodh pip packages from constraints to requirements","status":"MERGED","created":"2021-01-25 08:45:57.000000000","updated":"2021-02-09 11:48:08.000000000","submitted":"2021-02-09 11:45:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772259-osa-new-pip","meta_rev_id":"fbe575e87916c8e2b088f5ecf1e6ceea424d008a","_number":772259,"virtual_id_number":772259,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 11:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-09 08:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-08 10:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-09 08:36:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 11:16:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 16:40:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-08 10:07:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-09 08:36:38.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d40cfe9d5dea6e788e048087fea39567ae657c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-25 08:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"114bf24250dee09060a1473522eae23aefe6dbe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 11:16:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c4f5046207de43b6ae39655173ad8cf5 : SUCCESS in 9m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/aa47dc70d59d4e139ab831b0f47f409e : FAILURE in 1h 07m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6ab7f738aa6547448f62d0288b3751d6 : SUCCESS in 1h 38m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfcd7d5ab79048e2a16cafea346bbe5e : SUCCESS in 2h 12m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a06a8d88091439b9d34d51c39a04d91 : SUCCESS in 1h 24m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/10688bebc6294938afa710b4c1c07b94 : SUCCESS in 2h 22m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7478300ae24b4f988b414acd66df2376 : SUCCESS in 1h 55m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6e15b6c227604558afa25814486b33e7 : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5dd52956d4f4a70ac14db9ab6ac5813 : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7131647125de46c398380311fd05d583 : SUCCESS in 1h 44m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b1d82dfa102b4a219de5920c53af6a4a : SUCCESS in 1h 34m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/645b451f24cc4d5c976aa507ac2a65e5 : SUCCESS in 1h 19m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d206c3c89442c49b3f84e45962d3a4 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5df78db9e8c8dedc4ab07347a29e2efdf084f254","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 09:33:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"002c8ab858f82952652e1ae6f2f48cc4bfb0d2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 12:14:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb13151a98ee467cafae45fe13c5544c : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dbf288902c27437c9142684bfb1a4514 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5fc5014ee47c42ed989feaf8b63fd909 : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ddc664c231d46e2816630c88a37e0bf : SUCCESS in 1h 36m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86a96ba5eb8c482784c62d6cda272931 : SUCCESS in 1h 17m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/17bcfdd297e241a29d05dd774caae1ea : SUCCESS in 2h 27m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d80347b53f47b5858284307776932e : SUCCESS in 1h 52m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9425c16ce69426987db040cf142c5a9 : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e95e3d2730b44e0950ecb32194b973a : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0c6ea81b8fb74000b4ba79633ed2430e : SUCCESS in 1h 45m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f4b33830d4494e66ac82d733553ccc71 : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27a36fc2a19c4eaf868518cc25819c51 : FAILURE in 26m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f6a827d0e84319a8a1316cf8ec2a6c : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8e9b3705aea29c537c2ce9489f4d248d338fca71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 13:56:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d72debc214eee69642407a2b16fff9ca5cfa89b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-28 16:40:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aa5c3c67abea050895303050ca66309a8a0efd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 17:28:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f6adc426e04c442abfa2fed0c95448f6 : SUCCESS in 7m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f69854ba2279416f91e67a4952a41242 : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f48600a3aa844f3da885da308712605b : SUCCESS in 1h 50m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25c9f15f690a4e2e90c0bcfe8f4a17f2 : SUCCESS in 1h 36m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f165e63d134e44f68e3ca8a2415580f3 : SUCCESS in 1h 15m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/632820c5cd954aba8b393f831cb36c26 : SUCCESS in 2h 23m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0031f9ac19f4fb6b4317d6e9d369e60 : SUCCESS in 1h 56m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ad7ac84fbc14384bf11755385e320a0 : TIMED_OUT in 3h 03m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517b913fa1184c91972750cd2d7622d3 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bd30d80b5e3b41d58ce3c015f51eef7e : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4cd076003fea4e798c4aea30cea4bf00 : FAILURE in 58m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba236380bb64ca4a9f792b9258a3a7f : SUCCESS in 1h 26m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3daa2c2e66d54dda9c41f91392c245a1 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"02d06b2e41425514cb33aaaedeb2a9e7186cce80","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 17:32:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42703c916757401d42b2f2b0f2db3c927674512f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:47:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/acac7d2f120b48c89bf7591c27651319 : SUCCESS in 12m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2b20af824782450b9960d1f150f2deb6 : SUCCESS in 1h 15m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ea67380113b42a3b72aa26599b97d1f : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66fcb7519920486e9c7a4e2435320053 : SUCCESS in 1h 54m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae7558a103a14435a726071fefdf2a25 : SUCCESS in 1h 16m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2c937cbfb26d4b4c8851af0beb624be4 : SUCCESS in 2h 32m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1bddebf862d47478b9d0584b8f6e514 : SUCCESS in 2h 02m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f68c993a9cc54de9ae5511770d5b3012 : TIMED_OUT in 3h 03m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ab857f1e7bc4cc3bb08dd7fc8139709 : SUCCESS in 1h 36m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2f3eab21c7054e63907d2699470fc3a2 : SUCCESS in 1h 47m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/620a3e20a02a4f5eb3ac62ff7aaf12de : SUCCESS in 1h 32m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eb6cd79c5e0422c9f26a1b8943756cd : SUCCESS in 1h 36m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3452b08d7649459e66754a4da5d5b7 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c0d1ed35c6e3fd37650356abd1e549a5bac79312","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-05 17:23:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfe49da7f0f8ad76070159a7a25f0a634a1f0864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:38:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22ec12b181f24ec29d5b8d7b4d9492c3 : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bcf194dd90de44ac8379ce0b6083e808 : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b2c64c7fc0714ce481ddca89c4ae0c17 : SUCCESS in 1h 46m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0d78b00b14e4547ace6fe5df06a8649 : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/834e8990bd794ac6aeff7389cae7e9f8 : SUCCESS in 1h 18m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b5725fd9e18e4f709efaee1ebcef7418 : SUCCESS in 2h 43m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5647d768cc67451da0ce8ff796e12845 : FAILURE in 1h 57m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d36f92febc694775afa168770275fb5d : TIMED_OUT in 3h 03m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129f271f817745e6bf49a4815021bd45 : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ab1abba77aad4e9f84a6ba3622c5659c : SUCCESS in 1h 48m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad65c2aeb5474375ad711496749ab1d7 : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e28cd8b41ef84b97b882d812017c02f8 : SUCCESS in 1h 33m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68375d1ab904dec9c7fcc1ee29a24fd : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9ce53d0da3634a61ac155ee35daae5b361cfefc5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-07 19:39:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1015daf989c90f0fa3d25c28307aefa102083b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 22:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/79d97f5a4f8448158cfe8c65301d6b48 : SUCCESS in 8m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3da2ab2416ca4735bf57d961d615017c : SUCCESS in 1h 21m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e628beeb8ed48b086ab21fb75b5604b : SUCCESS in 1h 50m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/421c3a044afb4e1485ecefb85186b6d7 : SUCCESS in 1h 43m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc48f9073b247a0aa8f489be039aa7f : SUCCESS in 1h 22m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/425f8ab84d304ee7ba77876953c53f84 : SUCCESS in 2h 25m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fd0bf3e5dca480887750befc5526384 : SUCCESS in 2h 15m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f29fa8917e0844ad8d4125b1c11ce414 : TIMED_OUT in 3h 03m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b04ce58aa18e415888626c2450a1f051 : SUCCESS in 1h 07m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4f2db55569f84127b307ca2345700b22 : SUCCESS in 1h 53m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e517c5043b4343bf97953e9b5fbf1b16 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15d6f19100b64395acdf78c2e7aa2a54 : SUCCESS in 1h 21m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab6c97bf191478885df7af12b3ebef6 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0b6fab049b65d851eb1cef9433205af3cbbd9d70","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-08 10:07:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c399c7d5f464eac17a4a00b08c7f91125676782b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-09 08:36:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60f631291630861edab9cb54948f167c8bcc9a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 08:36:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fd10e120475384d99b9ebc34eef2b18ad08ff0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 11:45:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/625db43de7c64cf4bbe981df8c71f703 : SUCCESS in 8m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a0f431fc141f495bbef432f38cea39c0 : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/be88129c3c3446d7a2f81ee36df34f6d : SUCCESS in 1h 56m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37125e495c3143c6afdcaefd2b66a08c : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5477adee30a34b3eab2ad4e672aa0c58 : SUCCESS in 1h 18m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/69111c4f60824294b5c50c37e3815adf : TIMED_OUT in 3h 02m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf0fc5b7b5b1410bbd1e10cbf5dbd151 : SUCCESS in 1h 03m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4906ae8719d949ea8787f4076f0880c3 : SUCCESS in 1h 40m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8f0c957fd30e416e8311260433b90917 : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea4db7e56b794210a02fa870c9e25fdc : SUCCESS in 1h 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4da3b5b26294f7aa01cdc39c7fe15e2 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e30a7ca6d7e659c433b7267a992ad04c0fe603c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 11:45:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fbe575e87916c8e2b088f5ecf1e6ceea424d008a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 11:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c836cf997c40f888b3e22f877e3dd7 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"405f551fad8e08b1610459f30baa4101b92d00fe","revisions":{"405f551fad8e08b1610459f30baa4101b92d00fe":{"kind":"REWORK","_number":1,"created":"2021-01-25 08:45:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/772259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/59/772259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/59/772259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/59/772259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/59/772259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/59/772259/1"}}},"commit":{"parents":[{"commit":"d6d0cd22ed0b35f787ca135d6d3ced4d129244e8","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d6d0cd22ed0b35f787ca135d6d3ced4d129244e8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 08:44:38.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 08:44:38.000000000","tz":0},"subject":"Move aodh pip packages from constraints to requirements","message":"Move aodh pip packages from constraints to requirements\n\nChange-Id: I73773536938dff9f768381a333c5fa02c98e9d55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/405f551fad8e08b1610459f30baa4101b92d00fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/405f551fad8e08b1610459f30baa4101b92d00fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
