)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~779666","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ic2aab8aa458077d9bc778de3affb18b6d192d2bd","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa-default-python","hashtags":[],"change_id":"Ic2aab8aa458077d9bc778de3affb18b6d192d2bd","subject":"Switch default virtualenv to python3","status":"MERGED","created":"2021-03-10 09:14:20.000000000","updated":"2021-03-12 16:02:07.000000000","submitted":"2021-03-12 15:59:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779666-osa-default-python","meta_rev_id":"cab4ae131ffd97c4e395885a50704a8d4b11132d","_number":779666,"virtual_id_number":779666,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 15:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-11 23:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-12 14:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-03-12 14:12:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 10:47:13.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-03-10 11:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 13:39:37.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2021-03-11 08:40:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2021-03-12 14:12:39.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"db532dab050150c904ab95aaad63771c709f74a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 09:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2de7410cb5f53f8ae6b9dd3f77337fa503bad52","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-10 10:47:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"20b0ace54c8988f662251c9a5de5b730a5eab56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 11:23:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2f2c614b3d28424f95d0266805c3c605 : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8b6f4e0f766e41f79b36d1fe294a3fe0 : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5591d866ff36446e81a448b006c3b9a4 : SUCCESS in 1h 12m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cb9faf389a64e56bb2823015e1ec9cf : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4373beb39854d88b38fc1f5b2c99e8b : SUCCESS in 1h 17m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a5acb3418a8d4f1a9cd8af58c42a67e9 : SUCCESS in 2h 02m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0caf4a47dd1450d9a7cd8d333766d17 : SUCCESS in 1h 57m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7b00be0e0fc840f99dd7355bb8f60afb : FAILURE in 37m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb6a6a41e6a8427c91b3215d190dfadf : FAILURE in 58m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/61c455231b1b41ada4ee32762a7c809c : SUCCESS in 1h 47m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/88b39b5fdcf84e7eb25d25f345a8518a : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7ddce1c907a4a6f82b2f01402c35c87 : SUCCESS in 1h 16m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ff42f684d343959f3212cd95815146 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8c9a7a109a04a2c8c24a249748bb0dadb8f8200e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-03-10 13:39:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e7d98694617a2233a8068a10ac9f617d77128a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:43:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1be7cbc64d547c6ac030ad17bd363c3 : SUCCESS in 12m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1c7aedacf54b41079762ecbe74c105f3 : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/657ffdc6609148b7b7e698a7ff6c607f : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3181f99414a24338af96415bc3fb1b1d : SUCCESS in 1h 34m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b7c43f63dc74ec1926ffcc65df9f3e9 : SUCCESS in 1h 15m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3a1091cc8c374d8e9b0b4a68bbda5c3b : SUCCESS in 1h 44m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8d3786d0c904ba891952ccf20ae0a61 : SUCCESS in 1h 30m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a9e23c3be2e7492c91207c11c5d940e6 : FAILURE in 34m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e89b7fef70e0440da39396624cd72aa4 : FAILURE in 45m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f488e5809cbc483995ce1d44d20eacd0 : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5aecd47f98f345e3b685f18bbdf4f1be : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/255b177dea7d4d289b3ef533b57d3e2e : SUCCESS in 1h 19m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa54e98003e4da9bd076a230d6e2f44 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"53bd503c3089431d22520a2d5c807c0f28f037ee","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 18:40:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77a6e3faaa05f8af1a325830a057064169bd68df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 21:06:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db5eee59c33a4cbe8ba7c31d297b7b5f : SUCCESS in 9m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/faafdd9438814af4b9006f9eb311f3c6 : SUCCESS in 1h 39m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1ba6716dc2ab4265b8f44bf5153e9eae : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbb17ed1c3ec4e9b9d90cc16262a4154 : SUCCESS in 1h 33m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52e4a1dabfe349cca121de60da6aa5ee : SUCCESS in 1h 31m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/54f393ffee764d8692eac09bbcd8e88d : SUCCESS in 2h 12m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b44bdb04255446c88ca7eac07c1e620 : FAILURE in 1h 08m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c037fb1df962447db722c748cdd3ca40 : FAILURE in 53m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3068cc89092a413bb0d8a60f44987dd6 : FAILURE in 47m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d515da29ee6d448ba8c7ff447b27f1e2 : SUCCESS in 1h 36m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fc474f986c1047aba6ad954487f4598b : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e08d28baf984b58b9a39d879b29c3e0 : SUCCESS in 1h 04m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e49d762ed74b3eb83f94adb0f4ffdd : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e9f783e3327415c5987ffe959a590328e1c50622","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-03-11 08:40:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88d2b82e231e4c6488076a3c3bc5642682a2326b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 10:58:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1562d4c63a7458baf5b0931f2fe0a8b : SUCCESS in 10m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6fdb1de197514afab23e0ec32e8cc1a6 : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f01ea530e8e4420eb86c2d08461ccba3 : SUCCESS in 1h 36m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4838ee5ab8ea465ab75b34eedbb7c574 : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff7889c1385c4a43a77e92e9ae82f652 : SUCCESS in 1h 17m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a1fe98fb0fd54e69ad40b3bb35650035 : SUCCESS in 2h 08m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/782dc189dd16414eb3c6ce30b46f4399 : SUCCESS in 2h 01m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/229dfb95a55243939000227a6519a59a : FAILURE in 56m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0007d2335e4d4dd99c64d9e1740dedb1 : FAILURE in 54m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a46eeb43fcac4d6ebbfc4f02ea1c7537 : SUCCESS in 1h 31m 42s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ead4e790052b4f1bb160db5cb6d6aae3 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9fc8fc34f5a47e4a3fa6813461acd9e : SUCCESS in 1h 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da88b2c9256146899a7477943b1ebd11 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bf0d3da60b6ae6a994fea13c32a86a9d22cc8f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-11 14:37:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b52e4defae8d8fd813379d109e8985ac569e3537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 14:41:08.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"6eda9d2460853cf205af8718f7d190ebef523d55","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-11 14:43:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"49629a2dd9cf76daba8cc5e9121761a039d12ae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 17:11:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/73d1021fc53e473ebbca238ad6ab047b : SUCCESS in 11m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ae4d49d8989b4fd1ba7f104b446d9283 : SUCCESS in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7bfde2d5bc6d43a39e28efc0f92c098b : SUCCESS in 1h 32m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/469319b197e940e2a827eec1bcb169e6 : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b485255b196460b9035135b22729195 : SUCCESS in 1h 15m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fd88fc9c5ea34becb93646ce4740248f : SUCCESS in 2h 14m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd305216f3543e39df9a74bdaedd72e : SUCCESS in 1h 45m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/606a361cfeff4358a3482eb3c2c51c72 : FAILURE in 40m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be6afda6a404a3e8b258fadc1f0068b : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/50dda75017e04a07a126035fe056b111 : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3969f81d9d884a1e9bff625bd3f0807b : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb53ba3609ed47a6a1721ef26e031319 : SUCCESS in 1h 16m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2df531bd274597a4c204363269f2a2 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e1a4f1cdfc34023bdb2b39f9108fd02151b5315d","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-11 23:14:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a6553ae2edbf018d932109518dc7e9c33c64780","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-12 14:12:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6d50ca36984cf1b5be123d35ea741ebc13a31be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:12:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1747b16efb0613b422b3343a8dd1497a85dfd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 15:59:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e9e170a168b6486f964cc8e031daf35f : SUCCESS in 8m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3a180c2580014f8783c01c44bab9b97c : SUCCESS in 1h 10m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ebcc5069615040f4bd496af35f4f8011 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b71c460e45f43fb9c489a282cc551f5 : SUCCESS in 1h 27m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/322f7d9b5cf2453a8381089754847ca4 : SUCCESS in 1h 14m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b049913c9f3a40daa2e7b8f2bae6f9be : FAILURE in 37m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b916dcfd5f4a65baeae1ed24c18a20 : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f978d039102448fdbaeef4c37e3bb331 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3ed9f4cf26a44a5f84f560dcb4bbc88e : SUCCESS in 1h 15m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b6ad064938a42dcae4801e6a35b4f72 : SUCCESS in 1h 14m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcae984f63484467acb8b5c3a78f7024 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3d0230e0081f9d195336d8bb119596195a87ec12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 15:59:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cab4ae131ffd97c4e395885a50704a8d4b11132d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:02:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b5c3fd7db84765b8e09ed248c83b38 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"19eac02395ed713c418eb0d3f99ef231f92f2179","revisions":{"c1b6307793f2664417973075c584b14e79836aad":{"kind":"REWORK","_number":1,"created":"2021-03-10 09:14:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/66/779666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/66/779666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/1"}}},"commit":{"parents":[{"commit":"c6392f979709ea92473c56c1fe2f47de0d1cb351","subject":"Merge \"Move aodh pip packages from constraints to requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c6392f979709ea92473c56c1fe2f47de0d1cb351"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 08:38:26.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 08:41:05.000000000","tz":0},"subject":"Switch default virtualenv to python3","message":"Switch default virtualenv to python3\n\nChange-Id: Ic2aab8aa458077d9bc778de3affb18b6d192d2bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c1b6307793f2664417973075c584b14e79836aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c1b6307793f2664417973075c584b14e79836aad"}]},"branch":"refs/heads/master"},"19eac02395ed713c418eb0d3f99ef231f92f2179":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-11 14:37:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/66/779666/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/66/779666/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/66/779666/2"}}},"commit":{"parents":[{"commit":"c6392f979709ea92473c56c1fe2f47de0d1cb351","subject":"Merge \"Move aodh pip packages from constraints to requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/c6392f979709ea92473c56c1fe2f47de0d1cb351"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 08:38:26.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-11 14:37:09.000000000","tz":0},"subject":"Switch default virtualenv to python3","message":"Switch default virtualenv to python3\n\nDepends-On: Ia88b8e606a19ad8ee20665e125df1079b0a771ff\nChange-Id: Ic2aab8aa458077d9bc778de3affb18b6d192d2bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/19eac02395ed713c418eb0d3f99ef231f92f2179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/19eac02395ed713c418eb0d3f99ef231f92f2179"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
