)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~809704","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Ie02010583c9c9fc82211065cedac4fc120a31318","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/galera_pki","hashtags":[],"change_id":"Ie02010583c9c9fc82211065cedac4fc120a31318","subject":"Refactor galera_use_ssl behaviour","status":"MERGED","created":"2021-09-17 13:54:15.000000000","updated":"2021-10-06 11:25:51.000000000","submitted":"2021-10-06 11:23:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809704-osa/galera_pki","meta_rev_id":"e015e044f41a892193b3ef67667c5bf2d82fd8a1","_number":809704,"virtual_id_number":809704,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-06 11:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-10-06 09:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-10-05 17:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-10-06 09:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 16:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 17:45:48.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2021-10-06 09:11:19.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"bde52b29cc6aecc465136faeb551ee404663cc65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 13:54:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"876e9b96caf99bc86abce04d7af65400106b3741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-09-17 13:57:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cbd4ef46fabd2fea5a87a52fa3b0a9c4988a7678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2554facf36ea45e18791dc2a652610c8 : SUCCESS in 22m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4619c2e0c9d340759f45c85e699a112e : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/043a333dea4445bdbd5dd455632541d6 : SUCCESS in 1h 25m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9442ec4ce0274a1eb46e48cf3d1d5a5d : SUCCESS in 1h 34m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4432ddbc246493fbac86cb1d49cdf18 : SUCCESS in 1h 37m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd9848257a4f443d886acb997f9f0c94 : SUCCESS in 1h 34m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48b2927b07ea4446a3c57e39326eaa26 : SUCCESS in 1h 26m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5633c3f3e1d14fa990e7b00d09313e8e : FAILURE in 2h 29m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3187028d97334952b5d362590e01f07b : SUCCESS in 2h 07m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/73835053dd104593bc92728c123b3173 : FAILURE in 45m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a699eca59a3543f78735b4b06c8d17f5 : SUCCESS in 1h 07m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e937a752f6da454facb9de2646eaa46e : SUCCESS in 1h 38m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/11deb67fe48547e2b9f0dddf634c1693 : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/35753fda9e224786957d19e929fb18d4 : SUCCESS in 1h 37m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cf66697b79e34a2289436d6fbc904764 : SUCCESS in 1h 15m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1437d0dfdce445f878c585b11a77893 : SUCCESS in 1h 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d06ca2d6f574164ae78fca0503c8bf5 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b13330f7a0e2ef92113eaeb422a00c2888cd1871","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-10-05 17:45:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"50e2f201e4b423fdca518a98b1640a6ec0583c20","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-06 09:11:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f37caaad8e5064db3b8c301dd885c11ce9b2a281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 09:27:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"333219298f2bb65d93db0da8a80a7219934f14a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:23:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be3477c9c4ae40698f52a5feab166652 : SUCCESS in 19m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0c1116bf581b4d39b227ef6084402cb4 : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/de2ec4766f6f4b3d950739dd70f6dc3b : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/aa48e392dca447ad8707f2bf80f28841 : SUCCESS in 1h 36m 49s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d5622331a4344cc8e8175ca68b40796 : SUCCESS in 1h 37m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ec272340cc34a86a2b5ab4d0f0d3b51 : SUCCESS in 1h 34m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6efc6bd46ec34b6d8a3a7e8b49d42950 : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5396c9d6d7f46d9865b410a800d5a86 : SUCCESS in 1h 03m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9874765e8f004ecdb05839840263b3ce : SUCCESS in 1h 30m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1794613b965f49efaea3469f634ce424 : SUCCESS in 1h 36m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a9ea81daaf7743abba7ea2b0149104e9 : SUCCESS in 1h 48m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a5b5179463a54f9f86e02b2afd70f6ce : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3d8976091d4dd7a97b854523baa617 : SUCCESS in 1h 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7fe37febdf4e7988cbb8c5a623a99c : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7be5e4424135ebbc046bcea000cbe19d4aeb7d69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:23:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e015e044f41a892193b3ef67667c5bf2d82fd8a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-06 11:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6e90fb6a7c4a0c8339e9bbc753f018 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc541db1a2f314156f2116ecb1df556ac9223282","revisions":{"782546be36f6f6069e0fffb9456c2a66dda25280":{"kind":"REWORK","_number":1,"created":"2021-09-17 13:54:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/809704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/04/809704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/1"}}},"commit":{"parents":[{"commit":"bcedd4e7c22e6c3061c433c272bbd138936c961d","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bcedd4e7c22e6c3061c433c272bbd138936c961d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 13:54:04.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 13:54:04.000000000","tz":180},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: Ie02010583c9c9fc82211065cedac4fc120a31318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/782546be36f6f6069e0fffb9456c2a66dda25280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/782546be36f6f6069e0fffb9456c2a66dda25280"}]},"branch":"refs/heads/master"},"bc541db1a2f314156f2116ecb1df556ac9223282":{"kind":"REWORK","_number":2,"created":"2021-09-17 13:57:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/809704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/04/809704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/04/809704/2"}}},"commit":{"parents":[{"commit":"bcedd4e7c22e6c3061c433c272bbd138936c961d","subject":"Changed minversion in tox to 3.18.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bcedd4e7c22e6c3061c433c272bbd138936c961d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-09-17 13:54:04.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-09-17 13:57:47.000000000","tz":0},"subject":"Refactor galera_use_ssl behaviour","message":"Refactor galera_use_ssl behaviour\n\nWith PKI role in place in most cases you don\u0027t need to explicitly\nprovide path to the CA file because PKI role ensures that CA is trusted\nby the system overall. In the meanwhile in PyMySQL [1] you must either\nprovide CA file or cert/key or enable verify.\n\nSince current behaviour is to provide path to the custom CA we expect\ncertificate being trusted overall. Thus we enable cert verification when\ngalera_use_ssl is True.\n\n[1] https://github.com/PyMySQL/PyMySQL/blob/78f0cf99e5d5351df0821442e4dc35c49a6390c6/pymysql/connections.py#L267\n\nChange-Id: Ie02010583c9c9fc82211065cedac4fc120a31318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bc541db1a2f314156f2116ecb1df556ac9223282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bc541db1a2f314156f2116ecb1df556ac9223282"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
