)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~834845","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Idbc68f7496fd57f98fc77b9d0e345e576b51d108","project":"openstack/openstack-ansible-os_aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-04-06 08:48:17.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"Idbc68f7496fd57f98fc77b9d0e345e576b51d108","subject":"Use openstack uri proto","status":"MERGED","created":"2022-03-23 10:51:31.000000000","updated":"2022-04-06 13:16:43.000000000","submitted":"2022-04-06 13:15:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834845","meta_rev_id":"4cf39516c3354da7346ee48670232ef6d1a51c8d","_number":834845,"virtual_id_number":834845,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 13:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-03-28 10:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 16:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-03-28 10:30:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 13:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 16:02:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-03-28 10:30:22.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"2f5088d8dfb80c836890ca7fb8ad196a7dd4d3b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-23 10:51:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a12d522952cb597d7b9b8c08ac32e2cc0c59884f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:26:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acdebf27a804539a1d55c345875b4d2 : SUCCESS in 18m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f59a79c3a5ba421780f4b1e6c67f642f : SUCCESS in 1h 30m 21s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6261627593a74c2bafe535f176ab5ead : SUCCESS in 1h 52m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d925b9201db6482db5803abe852973a3 : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2a9cb497d81349f6aba33f2c39a42210 : SUCCESS in 1h 40m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc41a85e84484056986f86acbe2de9e1 : FAILURE in 1h 15m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/628a3a87cf4e474b8947acc0f7a1c3ed : SUCCESS in 2h 28m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d29ba46a2b694600acfd22c3cee89b7d : FAILURE in 36m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813ab9c54f9a45d6907a291f3c463444 : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c975b15fa1e64dd99f05a953389112f0 : SUCCESS in 1h 46m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6759c554de4d4fbcbbaa12efeb0a1f05 : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d5f835651d3b40fda11f395b435180c7 : SUCCESS in 1h 40m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/537f9882899d42b2b466fe710320feba : SUCCESS in 1h 38m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e96800fa7824a8eb861f47e13106b38 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5605990e9c183c585ada5210d389e337bd9941ed","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-23 16:02:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"248a761de5db19577d5d5fd6c4dcb85a18b687ec","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-28 10:30:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"257ccc7e0f21ab729ab11fb2f724d05a41129f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 10:43:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a9ddbd5645608684dee77158c7b993e658deb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:37:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f554ae536ea456b9eb0cf2ed5a5e46a : SUCCESS in 17m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/220b513c6bc8465eb33f7e1f7820d779 : SUCCESS in 1h 25m 26s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f215fbdba8f941b6822a7afb9c812b10 : SUCCESS in 1h 50m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ae980339cd4bff8ffa87cd4e984f34 : SUCCESS in 1h 41m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/66b7732586fd4eb480f768e506decbba : SUCCESS in 1h 46m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5cc91efff5e42d0af15e48d5f4ef350 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbf9977f5baf41e3b3ad3ee061a70e7a : SUCCESS in 1h 51m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/11d06cb916644e8c85e54bd26b6bf1de : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/afaa3b673d1448debb913fd4835a410c : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aca9a8354be24fd19b506fe4a330562c : SUCCESS in 1h 38m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af29c3f0571417cb810bf4307690cd0 : FAILURE in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9cd28dc83b5b7ebb050a5b78c27275a00952f06e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-06 08:48:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b74c20a04ead31a474409993e6c8fede80d8489f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 11:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0107eb950d440cf8ee75f6e481ddcc5 : SUCCESS in 15m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cd69f28a74a64e6c87026bc4311cbeb7 : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/def2fa915a3346d8b72c99c3231fde6b : SUCCESS in 1h 40m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bf20c83c0a84a4ebaec7b5e95a0202d : SUCCESS in 1h 41m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4d9e518e428d4595b345744c95ea87c2 : SUCCESS in 1h 41m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb7160647dda4d38b2661034ebf77dc4 : SUCCESS in 2h 24m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d5df1c8b064890b43ef5d64f429dd6 : SUCCESS in 2h 26m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e55a2d205a54232af49cf10899bd09a : FAILURE in 34m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad9984178274426f8e6fdc3d0a9e0895 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/29b00128223e4abbbed9c35db9857a57 : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ae42d6415e8b486eb286d76bddebfb41 : SUCCESS in 1h 27m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d215b6ae70b1411eacca2b9c8918c8e8 : SUCCESS in 1h 38m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217c9afbec4c4aeb81c82f6050e3ff79 : SUCCESS in 1h 36m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b5c3b808fe4d318053d1ecc8b9faf3 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"61993fd53dcd2fdc90aad7ba6d164599e20d0bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 11:21:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24004383744ef2ad3f9d18af49580bb037b55e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:15:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1afe17080a794bdcb652f868c0dd30b5 : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cd182806d193442ba6fc00d89eaf2b8b : SUCCESS in 1h 30m 32s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdba5d7a365447f4887216a8c3242ac9 : SUCCESS in 1h 48m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c00de5132a254040b71edee6224a2c2b : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d3258ee293344180af28cef89e0bd2a3 : SUCCESS in 1h 41m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd425be6980c4781968f208e4bf8a182 : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba073821b86f4efeab6606053c3dab70 : SUCCESS in 1h 41m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2e39f273a04f4c448d40095965f835cc : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/eb170fd73de74b09b4786007a1dbcc66 : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e14afcb96084d1fa3f884be33efe439 : SUCCESS in 1h 40m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8850b724a9c54ca381e0543b02a85d6a : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d6cd0b86c8f17bc2a42ca4cee022e7fc8a8b21a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:15:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4cf39516c3354da7346ee48670232ef6d1a51c8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 13:16:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec264d3d36b540c685ee08c73373ab63 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"426737ba6a70e1fd45f498a80aea97023dc84480","revisions":{"426737ba6a70e1fd45f498a80aea97023dc84480":{"kind":"REWORK","_number":1,"created":"2022-03-23 10:51:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/45/834845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/45/834845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/45/834845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/45/834845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/45/834845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/45/834845/1"}}},"commit":{"parents":[{"commit":"2e3e39ab6b28425c7dd6c00ec761673e74f2d3f6","subject":"Remove legacy apache variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/2e3e39ab6b28425c7dd6c00ec761673e74f2d3f6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-03-23 10:51:23.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-03-23 10:51:23.000000000","tz":60},"subject":"Use openstack uri proto","message":"Use openstack uri proto\n\nAlign aodh role with others and use\nopenstack_service_\u003ctype\u003euri_proto variables as default ones\nto define protocol being used for endpoints.\n\nChange-Id: Idbc68f7496fd57f98fc77b9d0e345e576b51d108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/426737ba6a70e1fd45f498a80aea97023dc84480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/426737ba6a70e1fd45f498a80aea97023dc84480"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
