)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~835698","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I49522101226fd4ff9cf3b93ee75855f07a9574c3","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","attention_set":{},"removed_from_attention_set":{"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-04-01 17:13:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49522101226fd4ff9cf3b93ee75855f07a9574c3","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2022-03-29 17:17:50.000000000","updated":"2022-04-01 17:15:32.000000000","submitted":"2022-04-01 17:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835698-openstack/openstack-ansible-tests/sync-tests","meta_rev_id":"b7fb158be044aeed3a65c93a3be96a7d63a1b2a7","_number":835698,"virtual_id_number":835698,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-01 17:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-04-01 12:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-01 14:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-01 14:26:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 22:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 12:12:01.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-04-01 14:26:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"13b5f503dac31ade207d9e54bce9c53a0ace5246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-03-29 17:17:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bcba19e1ca3cba24b2de6975e49f464acbef98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 22:14:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/721a59ec9bfd4007b2471c112cadade8 : SUCCESS in 19m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/51428a43dd4f482aa62b7d6f58f4694d : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab57f347518941b9b91400b1c26e0bce : SUCCESS in 2h 05m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/505c1245e15c4fa29f18680d3a3a4a2e : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d9f9d935de5947c4858e82d2b00ece86 : SUCCESS in 1h 44m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d25fd05e7504a7a9414848c3529349f : FAILURE in 1h 13m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba03df7ff884d37b6eee81bbb4547c8 : SUCCESS in 2h 40m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/824e8b7cded84e16b047ef87fcd8f5d5 : FAILURE in 29m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cecdf345748541ddb4c039b9dbf57995 : SUCCESS in 1h 04m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdea47f2e19b4a15ae6b32fa9e7ab366 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2ccb4e1488144a0e8d6ce62e350e8dcf : SUCCESS in 1h 24m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a6bd581f6c5844b8915b769194eb0f69 : SUCCESS in 1h 46m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61ca1297134543438f33d5e6511b186f : SUCCESS in 1h 44m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e6cc2a11ab4fe099257e6f4bb13038 : FAILURE in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"03600dbd4576d2ed6bd8963a7bbeca353a3dba7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-03-31 20:14:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ac98dbccb8e52e76710955a7576aae37e7cbba2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 22:52:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c44813dfdc49fdaf6d2fd629df819e : SUCCESS in 19m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9f02e0633bb2442b98f2c16ac4e41e67 : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a56cba34e70c4254a818aa23f83b22d9 : SUCCESS in 1h 44m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/733e68e454104717b2a87dd7abe4bc80 : SUCCESS in 1h 43m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6b1933fd33a9407da7af324a1ce36202 : SUCCESS in 1h 43m 14s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dece8d4da464db59458561ae79a943d : FAILURE in 1h 33m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/976acbc0998a4ad89a0f2c9ce47a1560 : SUCCESS in 2h 34m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/05dfe476770d406d8578bae0aa879dbe : FAILURE in 32m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcd0051eff3745bd9238869428b9f733 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8faae98fa8c54f4ea728a43656eb5b63 : SUCCESS in 1h 51m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a7f74df5233c459585551f49f6f8fb2e : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1860bbbd2f3c4828938ad05c416de7e3 : SUCCESS in 1h 36m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35c4616bd66e41c88b488b5ccc64208d : SUCCESS in 1h 36m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd3bb8fc6324d399e659ea913c747cf : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f20ca44b8332bae932839cca38ee63cde7be9e8e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-01 12:12:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e6fcf87da0abc0619f3099a2fe602a57a9058aef","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-01 14:26:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"51e5e5eb202d3b719f0ef0c6631c9cf4f4f74474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 14:28:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8abeb58c28609007afb640bdfd9eb03377addef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 17:13:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f10cbef8c8f40febb7144ecafbad05f : SUCCESS in 19m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b9efe951f57420daada927259b1a9ff : SUCCESS in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/64b89c0ec772460b828685a3b63f4010 : SUCCESS in 1h 44m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc17d4e517714c61bf4ea70d0e037353 : SUCCESS in 2h 35m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d91e5c5d99c14980b29862cf416f4679 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75310754e2e742c78abdf7f0d0e97b2f : SUCCESS in 1h 41m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/53526fe046e04091a555bea51f56b761 : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/31b1ff8853b8497fb9f89b1b3c5c1678 : SUCCESS in 2h 06m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e830ca0004d94df19ee963c43ae06494 : SUCCESS in 2h 35m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bffb78ea84104fe9a5b05fe729899b68 : SUCCESS in 1h 52m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75030d327731406496e914edcc566389 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8329babee6a727e4146b7042040428a442971ef5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 17:13:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b7fb158be044aeed3a65c93a3be96a7d63a1b2a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 17:15:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665da4f2335045f8b35ddf5a3d5c1d4b : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df236fc2a3cb19dd92ab3cbbc7bec2332741bfdb","revisions":{"772cd7d858a661df71beb43b3ee0f81d42cc9b88":{"kind":"REWORK","_number":1,"created":"2022-03-29 17:17:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/835698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/98/835698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/1"}}},"commit":{"parents":[{"commit":"2e3e39ab6b28425c7dd6c00ec761673e74f2d3f6","subject":"Remove legacy apache variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/2e3e39ab6b28425c7dd6c00ec761673e74f2d3f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:49.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I49522101226fd4ff9cf3b93ee75855f07a9574c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/772cd7d858a661df71beb43b3ee0f81d42cc9b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/772cd7d858a661df71beb43b3ee0f81d42cc9b88"}]},"branch":"refs/heads/master"},"df236fc2a3cb19dd92ab3cbbc7bec2332741bfdb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-31 20:14:23.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/98/835698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/98/835698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/98/835698/2"}}},"commit":{"parents":[{"commit":"2ea4e53c6f87817f758e0b67d8a04523c5b4e3e0","subject":"Cleanup setup.py config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/2ea4e53c6f87817f758e0b67d8a04523c5b4e3e0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-03-29 17:17:49.000000000","tz":0},"committer":{"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","date":"2022-03-31 20:14:23.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I49522101226fd4ff9cf3b93ee75855f07a9574c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/df236fc2a3cb19dd92ab3cbbc7bec2332741bfdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/df236fc2a3cb19dd92ab3cbbc7bec2332741bfdb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
