)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~882008","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2Fyoga~Ic56d5d1bd1ead8fd6bb642913370fa206799256f","project":"openstack/openstack-ansible-os_aodh","branch":"stable/yoga","topic":"cherrypick-osa/systemd_restart_on_unit_change-nq9phb886h","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-02-21 09:20:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic56d5d1bd1ead8fd6bb642913370fa206799256f","subject":"Ensure service is restarted on unit file changes","status":"ABANDONED","created":"2023-05-02 11:02:01.000000000","updated":"2024-02-21 09:20:37.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":879954,"cherry_pick_of_patch_set":1,"meta_rev_id":"8735af0dd725712ddb5f4bcf21a8316fdacdfbc6","_number":882008,"virtual_id_number":882008,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-03 18:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-03 10:52:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-03 10:52:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 14:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 02:08:18.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"CC"},{"updated":"2024-02-21 09:20:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5bd173ddb0ff4e32bfad09d9e82894f1d7d49883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 11:02:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1dbb46686dc763acf2fd627603e351a2f61fa2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:33:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81581b84e6e44b79acc859cc0169a8cc\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988072e62bc84fc0a35c25ada2952cb7 : SUCCESS in 11m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/920798105a874fa2b41096693d465f8d : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ed1774bcf6d463382bb71d1db9e4cdb : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/98f9b66df7a04eea8737dd663b4bf3e1 : SUCCESS in 1h 33m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e8575246fad403f91eed32b225785e4 : SUCCESS in 1h 34m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e126ef263724bef9529e3b867fb6562 : SUCCESS in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2b581ebe93e4275af548f7534f1bb9f : SUCCESS in 1h 30m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cf0ba14a71e44b849d72bc664835fc8b : SUCCESS in 1h 25m 23s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/95aacf0dca1d4881a807225da374ca0c : FAILURE in 2h 33m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/082f9f86ca814354bf67f4651b697604 : FAILURE in 15m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/77434556ed2a443583aec2648bb97ee3 : FAILURE in 1h 04m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcef83a7c3764aaa8eb1e1dca4fce71f : FAILURE in 37m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aa114ece9c54aa4845b464bf534a0a1 : SUCCESS in 1h 04m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/149693525c38472eb62be5be9dfa7d77 : SUCCESS in 1h 53m 19s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/24e9d01e81fa4284aa3f381c54f97ecd : SUCCESS in 1h 31m 43s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52a2e3c13a334dc59135d7a21c53c33f : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5a9e0494a068473dbf29a0754b51b6eb : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4dbbefb4e53b45bdb58cf055529827aa : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77cd583c1dcd4fbcbc34d17050f5d413 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9f2343c57ba495f8203a3e36ffc0169 : SUCCESS in 1h 24m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7bc3b5986e452eb888a41d4dda2143 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"60b0eadca952782ca890a6305cb62a057c6f194d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 17:43:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7792807ea22b5c74c546d35d732fbb30768d266a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:44:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebccb4093b1647b180e603724a2aa717\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e37f000276174e1d949c2f27b20d08a1 : SUCCESS in 12m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e65c8343684e4555b760856c84a95209 : SUCCESS in 1h 23m 02s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c630b6df25e04c989b6c04ffb9f2fb57 : SUCCESS in 1h 29m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5b10800636541408541c9c0c4f9e5c6 : SUCCESS in 1h 26m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/54e7b2ea0ff143fbb360ccd1c0264a26 : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1996af3b21c545a5b650b0217c34cb0c : SUCCESS in 1h 34m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33dfea98ed724a3ab1a6781a5370f110 : SUCCESS in 1h 21m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3c637622b9984a67b5754a67bd8c01ca : SUCCESS in 1h 54m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e7e460740694fd184263e66b16a0af9 : FAILURE in 2h 50m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf2143ed5b6f4fdca69f2155628f3e73 : SUCCESS in 2h 18m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/34c602ef2ccf4f39b2238584ac0ebba2 : FAILURE in 1h 18m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8e407bd90120441baf601788b1ecaa3b : FAILURE in 41m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dd08428fad24ea89a1d709539be4045 : SUCCESS in 1h 12m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/84fcf69f10fe49319259901483f8c790 : SUCCESS in 1h 56m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ae31270d921424ea41046f139a75a30 : SUCCESS in 1h 47m 45s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7c8110e0706470fab416d32231d1eda : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e27e3e360cb4756905b741a4d78d8ce : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/66fa85e5809f4d839b5bd1a8db39b808 : SUCCESS in 1h 45m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a18615e86bad4509a08f2f11667f81e8 : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfc23447a25d4f79912726c3db447af2 : SUCCESS in 1h 20m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee718a92082f45479b8bf09cb7dbfcb7 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0a8c448051b0a874cba906f0054f916c633a07dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-03 10:52:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"beefd402c7cb91ba83470b084c2779825c068978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 10:54:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44f9ff2d606053ac663cbbf9a15b0320dbb682e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 12:45:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41cefbc1b5df4664969afd400c693131\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc21e0bbbec44f18a82ee41eecec021e : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57212bb67c6a4ec1aed0871d95f18279 : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bb146bbc0ee4374bd9037260a2667a1 : SUCCESS in 1h 25m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/540e25ba3678499a9b9a0477159e7e16 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f700d51b55c94dca992a984d0aae2585 : FAILURE in 1h 46m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3047790393d44d9982a92a90aac8307f : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a877855430045cf84cd96ad810ee282 : SUCCESS in 1h 05m 22s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/99cde702a9e74ed3a02515fc1c32067b : FAILURE in 42m 37s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d04606e3be147249c8c0f412e03b9d0 : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d45508f7d1b4901a1eb38e0a4796fad : SUCCESS in 1h 36m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b04a565fd67b45138a45ea31ba669cbe : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/84da334cd350447aacbdf07b8b8f30e9 : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f923976ca824cf98b66964ae44c4205 : SUCCESS in 1h 45m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a943f427c44e43cdbbca14bd2aaab967 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"69c1c63829e6244331793ce235f4b19c57433ca7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-03 16:07:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ec683f2b6b923ffe3900d41ccc55368cd6b7494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 18:49:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/780fa1d1a23c46ddbfb06e35ff148de0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e6f9f0b6933483c9d0cf09546bbe161 : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/eb1ab7a540e7497ba4410bfa1df2d695 : SUCCESS in 1h 14m 37s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/df053fa08c574c619cc1be331bfbcb08 : SUCCESS in 1h 38m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d9054f914744d548a5075ff59ed06e4 : FAILURE in 43m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/249a1e3d327a4f06bafc9cbd21f1fbb2 : SUCCESS in 1h 39m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b53ea382a6a493b9226b62652337cd5 : SUCCESS in 2h 00m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/293cdfd3089b43c89da5ebd3d6c9f06c : SUCCESS in 1h 20m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4c4a4139f293420eb24335f1cbcfbefd : SUCCESS in 1h 42m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3d6905767e744449957b8bf514dc98f : FAILURE in 2h 34m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df5516e4d184be5aa929e06d7f659c0 : SUCCESS in 2h 07m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ee478064d3c43dea160a38fdf018e3b : FAILURE in 1h 18m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2f5692ed284440938a2a866d8ac87552 : FAILURE in 38m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f84389adc8a94b3bbf8625e2b4044444 : SUCCESS in 50m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d767cf1455334b6b9365cc46ca7d879e : SUCCESS in 1h 56m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d416948e242142cfba22667f9f4c5ec5 : SUCCESS in 1h 29m 36s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e2fcafdbf2d349de836c622b96bffd31 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dca02922e9344c269e61382ec479e16a : FAILURE in 1h 12m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2b632796e0c943c2a30548b72e1e96a1 : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c385f04123c415ba34731668d3e6d7a : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f3dc7a4ddc0483cbf992002c36c5f3a : SUCCESS in 1h 09m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b06bf187e54af7a301c5dfdb4b0718 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c047c73707c6ab6589748e064b598bf64814e772","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-04 10:45:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c75ebce31275663f5a2c6e4f881b877cb87379e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:46:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1239172944447f984d29b8679f4cd0d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d78fb67d96c644638b5d18fe8c37231c : SUCCESS in 10m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c81f8d95c7694e2f8fe24a6834a39408 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1cb9d16d43140c2ba10d2cc22595cc7 : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/53911e99278543ab824b9c67e3046559 : SUCCESS in 2h 02m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/baf5ed191bbf47cf91a026504f371c3d : FAILURE in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60e6019e9fe244dfb7992d6f0e5e19fe : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a25312d94f2a4bd09792abd452a0b5cf : FAILURE in 1h 25m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1c966a4bb00f4d528d4955fae49f05b7 : SUCCESS in 1h 43m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c47d773680bf402dbfab76f0cbd2c8fb : FAILURE in 2h 42m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76de29e37c3e447e8c6bafab87a34ba8 : SUCCESS in 2h 52m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83db1d86a5944db888f8caa842377165 : FAILURE in 1h 09m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d938a985f134ae285835ca8bc922f28 : FAILURE in 33m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dabb73d9d014fc3a328d5a0057ff76f : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/079265cf16b34332ad684babcb752a30 : FAILURE in 1h 31m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d77aa4e3e11f49be859f408ccdb0fce5 : FAILURE in 1h 38m 23s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1df741b7612d428a8dd159acea6c5203 : SUCCESS in 1h 30m 59s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1aab2db705d04f439d69ce4d7f297498 : SUCCESS in 1h 29m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/06c1e906195046c9a62c39b8c8e5ad83 : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9a808492a394053bc35c182054e12f3 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b62f5ea5f30f4872be619dcfa7a0a4ac : SUCCESS in 1h 25m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e456991f18274c149c1a8c3ea23e3c74 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"64d85d85dc088a23702ba993a2e1d73e7aef891d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-05 07:45:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc90b62615d2197925bed27494e559f167239b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 10:14:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb6cdf7542ad440baca3b1ea9637ddee\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96524e10f1c142c1a88802fb67202de8 : SUCCESS in 11m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cc0f05d775084fb9b3c0bdc941ba507e : SUCCESS in 1h 18m 09s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f8f155cf20942eba7ee18b2e61a2cc3 : SUCCESS in 1h 47m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bbe8a4fd2664c398a294d91269c9b47 : SUCCESS in 2h 04m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0dd38969850d4046bc0ba97666699fad : SUCCESS in 1h 38m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3a783fe80c4b08965d9c4f378eb372 : SUCCESS in 2h 09m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80c22387b0b74379a5c583d921dfa350 : SUCCESS in 1h 25m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5d71ecdbda0c466da323fc57b7be8ec4 : SUCCESS in 1h 30m 24s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/768ba3d2216b423798bf20bcdf1a12b0 : FAILURE in 2h 13m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a43a8a33bd4f2ebf7f8dc2d7452e38 : SUCCESS in 2h 22m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/21d3186e0ac043b7bcc96fa72e06d68d : FAILURE in 1h 28m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a236650e8a8e4964aaaf689a7ecec129 : FAILURE in 39m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25cd438f1bbd432c9ee0cd2adeef45b9 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a0e85c7f76845da9e7291dce08a6331 : FAILURE in 1h 25m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a49986e8ea14fa7ae302d0febc01466 : SUCCESS in 1h 27m 03s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d83940c4c26043e8b7f271b828f70c5f : SUCCESS in 1h 37m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7508824da36f4db49cba9dc924a7d649 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8a2234beb4e74d96a545f31a88a31114 : SUCCESS in 1h 48m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c5d8f3c68f04c5fb55a1a23b19c07ce : SUCCESS in 1h 38m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f96bbb2672744d35bed677291023ae40 : SUCCESS in 1h 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309d96607e124401a9a1f70a429e7e87 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"923d84e989e6589c123eafae093faca46976cbdc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 08:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0db7725b71cb967494795938b95abc172a800c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 10:23:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91e5190a19734065895bcf73ec0be7e4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e48932650c4e4ca5b79f4869cad3ee43 : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d4cb815ac2d143e0a82b362b46b38edf : FAILURE in 1h 05m 55s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d1188fddf5c4719a21b7459730027c3 : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/763b2f0af1ed4da8abfaab1be94dfeda : FAILURE in 1h 00m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06f75a43de0b4ee99cee909efcae7be8 : FAILURE in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/106d762389b44ffb8014af8fb51ce9f0 : FAILURE in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba9d691b9824d7399a2ca057017affb : FAILURE in 1h 07m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5a7576c2731e4576a92bebf347bdea27 : SUCCESS in 1h 46m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f37edf087b104d89b4bdd0189313fe49 : FAILURE in 1h 11m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dee7686c1f64a45a705a8c09118f72b : FAILURE in 56m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0886c4391e5b4da2907290cf473310f3 : FAILURE in 1h 09m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51c52ccd08f1467a93c07c6dc9875256 : FAILURE in 34m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f9a162ca91e4619b3677cb9a5ef4bae : FAILURE in 56m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3d7dcf66a784f09b4799684a4e512cc : SUCCESS in 1h 38m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2e6e016be4d4db982c1ec05e72eb341 : FAILURE in 1h 00m 50s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2d8d111d0d89478e911273cf314d4d3b : FAILURE in 1h 16m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/acd0a8b53a4a4a6988e0d84d3f003d1b : FAILURE in 1h 07m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c143c9900cd4493ab9fa6ebfe4be7905 : SUCCESS in 1h 41m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cecd5d64d6ea4a468f90c8787ce2ac7a : FAILURE in 1h 19m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f578b86aea0040d39954edf4daff55c9 : FAILURE in 1h 12m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc7e2a6a84d499485ebb72749a7212a : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a4fb439bb5ed58c6d3988e5e40831da3fe22f9bd","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-08-30 02:08:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f2cd2b54beda91bce37ee12dafc28614d1ab0f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 04:25:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8212cb68ea924e2dab14cc6d9cda62d2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9d22d1eb5e948ee91edf2e470915136 : SUCCESS in 10m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c8224c12e7934dbf9a3199a59ab4d9ea : FAILURE in 1h 11m 48s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0621e9b301154b2bb8ead9028cba787b : FAILURE in 1h 42m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f889662d9f844022aef22b6bee191cab : FAILURE in 1h 03m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7aff3bad3d354ee39a0d1452b6c01792 : FAILURE in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b158c813b1d64778ae9947faa6c6c0c3 : FAILURE in 1h 20m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/372e73d888334fe2bb46df7887dbe145 : FAILURE in 1h 16m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/eb253e07f21c4766af05ffa4f813561f : SUCCESS in 1h 32m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fede574ebc6444d905bdf943542e5f9 : FAILURE in 1h 17m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c18c971fe16f4fadb53028c28a12d5c4 : FAILURE in 1h 00m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/10c1aceb40814872a994cf94693e913e : FAILURE in 1h 07m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cc3888569a743f7b62eda845ccfea09 : FAILURE in 43m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7ea030c77a34941bdcd2dd48c3be037 : FAILURE in 57m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/829765f894f94beeac7692f578ae9f1e : FAILURE in 1h 39m 17s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e74248d30c148e08d9f512a564bdd88 : FAILURE in 1h 02m 39s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3a1903b05aa4c94aa40c76e12faa14d : FAILURE in 1h 18m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f4d1e40cbe4f4809b55b50b8df83bfe9 : FAILURE in 1h 03m 40s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d7ad66e8fd6b4d9a91cabce14493315a : SUCCESS in 1h 48m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d7ced1db84740dc9c276c1a43c34054 : FAILURE in 1h 17m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a6eb5ac6219457ba9a348e4c2de06cd : FAILURE in 1h 09m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37716de6829440019cc6fdbefb7fe98b : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d7688bc8dd66ab26ef722f3343e615b63ac5b1c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-12 07:50:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efdb4b17acc0a9124a05f2b05c0e189ccae10dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:56:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d544f0f4fdc4ab68fcc0cebe862e079\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c398e8f12a9148ea81c808f35f57662a : SUCCESS in 10m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/61d6c675f5ae4339a0af62665c4dabe0 : FAILURE in 1h 09m 52s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e1b1fa9fc044696bb902c09530c777a : FAILURE in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/499b9223b6ac472dae4b8285c2693254 : FAILURE in 1h 03m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dc32f3de23f44c6a926c94abf46d871b : FAILURE in 1h 27m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a983b283316d4ed6ac42c7c5601b9550 : FAILURE in 1h 20m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c026ef46398b400295fc34513263dade : FAILURE in 1h 20m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4a5f261e189549a3b7bb046ce3ea8e27 : SUCCESS in 1h 27m 16s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3a40d473b644478924d934c2102d03f : FAILURE in 1h 14m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0ec8bfc4a74999b35d47589436969c : FAILURE in 52m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/930ce33866fb46d68e54b7cd68f4e0b2 : FAILURE in 1h 07m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b394906992f4b5ba54be406c93fa5b3 : FAILURE in 41m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43639fafc3d84498918e00fb61e5fc7a : FAILURE in 48m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d12035fb20c1403a91fb8288c89eea10 : FAILURE in 1h 57m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5fcc50aed6b420db68e2e665e039ccb : FAILURE in 1h 00m 35s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6e4ba35b17c04e36879089099fef948b : FAILURE in 1h 26m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/86e59924e3ad470191162583f944ae12 : FAILURE in 1h 10m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b43280a1300449559bb06d9e240f4255 : SUCCESS in 1h 28m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a9130d21a8b41fe8e40f543e5119a3b : FAILURE in 1h 11m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff6f14d0a2674a80b283f2fe64eef20e : FAILURE in 1h 16m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80405b8417184709a9c5aa7bbcff393a : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8735af0dd725712ddb5f4bcf21a8316fdacdfbc6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-21 09:20:37.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/openstack-ansible-os_aodh is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16f151e89c3677a5f4d580df0437d5cf5a0309e6","revisions":{"16f151e89c3677a5f4d580df0437d5cf5a0309e6":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:02:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/08/882008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/08/882008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/882008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/882008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/882008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/08/882008/1"}}},"commit":{"parents":[{"commit":"1b128b64d21c4345b7088326796fb22b439dc84e","subject":"Update .gitreview for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/1b128b64d21c4345b7088326796fb22b439dc84e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 13:29:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 11:02:01.000000000","tz":0},"subject":"Ensure service is restarted on unit file changes","message":"Ensure service is restarted on unit file changes\n\nAt the moment we don\u0027t restart services if systemd unit file is changed.\n\nWe knowingly prevent systemd_service role handlers to execute\nby providing `state: started` as otherwise service will be restarted twice.\nWith that now  we ensure that role handlers will also listen for systemd\nunit changes.\n\nChange-Id: Ic56d5d1bd1ead8fd6bb642913370fa206799256f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/16f151e89c3677a5f4d580df0437d5cf5a0309e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/16f151e89c3677a5f4d580df0437d5cf5a0309e6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
