)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~888471","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I53c0166beb446802641c701f6e43d5e317f9d54a","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/core-2.15","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-09-06 00:37:58.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-09-06 00:37:58.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-09-04 14:21:50.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I53c0166beb446802641c701f6e43d5e317f9d54a","subject":"Fix linters and metadata","status":"MERGED","created":"2023-07-13 14:11:12.000000000","updated":"2023-09-06 00:38:59.000000000","submitted":"2023-09-06 00:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888471-osa/core-2.15","meta_rev_id":"8f9f8bcf0ccdbd70a7a3c39ede66b25b5092b5db","_number":888471,"virtual_id_number":888471,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 00:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-09-05 07:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-05 22:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-13 16:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 21:02:55.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-08-14 08:42:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-09-05 07:28:03.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"2c8fbaf08c6306a4ba9b784ef492b2f0c1b2efed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-13 14:11:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a724fff53225a4a12e36fdf98f5fa027cdc9b24c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 16:52:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad048d263730464192d7e16e0dbf8d14\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7bba8294ce04261ad63e60380e5342a : SUCCESS in 16m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c5cfd98a9e6a43be9da933a3d590fd47 : FAILURE in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bbf4fb7f22a4fea8570393e949dfd1b : FAILURE in 1h 42m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6413aa6597e4163a8d6324965b6ec22 : FAILURE in 1h 24m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3c73bd8e3654ed7a6bdf974f9f7c84f : FAILURE in 1h 42m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8826a821fe724e7d995ae3a31c20296c : FAILURE in 1h 12m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/047e7844805041bfbf5d19b255264ab0 : FAILURE in 1h 42m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf83e1a099a644d4a1c857a08d88107f : FAILURE in 1h 31m 12s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c6ac52c9bf8472499b4254811c75781 : FAILURE in 2h 33m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de526cbcd3f24402a8054622d567d9cc : FAILURE in 1h 19m 36s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd0fc6cfbb444ca889025585796f9e20 : FAILURE in 1h 01m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88c9e9d0415e43c4b97d8360972ccc7a : FAILURE in 59m 52s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d2d3d6f3e024d2fb171833826c79cb4 : FAILURE in 1h 42m 25s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba5cc1f5533f464f9989a83ae55234b2 : FAILURE in 1h 49m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/764d2228326a49b1a115c3b5fd851d51 : FAILURE in 1h 36m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70f3f92e22e34384a5b0ef4ecb470ee8 : FAILURE in 1h 47m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a7782fa2be432a8441205037606ca6 : FAILURE in 1h 36m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612b8480f9ee4d83b58adbf8f7778cec : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"924dcd62f6f44f19b2c4f9b7423b2cf68d2d5a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 05:36:40.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"14fb69d6ca678ab26c1ead8b273af56e54be186e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 08:05:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcbfe048bac7442ba760c31626140366\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/276c74b172924824a50a906229b373f7 : SUCCESS in 10m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/da03a7e1d85049e995b4a32c56f4827e : FAILURE in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5080c69202b64a8abe14353bc9784ff0 : FAILURE in 1h 34m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9bed7585a7954e3bb17fd0edfca9e74d : FAILURE in 1h 43m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c645ed7fd11412a8c820b3b4f88e1a0 : FAILURE in 1h 37m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/390623bcf6fb42669693c253efd906cd : FAILURE in 1h 22m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db399e98e74a46279b2215ab06f0f26d : FAILURE in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e085567defc4a44b60d564e000f156b : FAILURE in 1h 21m 58s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f983a0d6fd449c9b7d98c08f69bcb08 : FAILURE in 2h 20m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96e61ec2ad404b6eb5d7bad9be738362 : FAILURE in 1h 55m 07s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e58f51c1de346afaec7af3ba488d745 : FAILURE in 1h 00m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08aea1d08a474886a6f2e2315a7c6565 : FAILURE in 51m 24s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/63d0ce629a9c4fa3a51247df8eab8d46 : FAILURE in 1h 20m 12s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/beb8138725384ef7ba544178d4bbc4e4 : FAILURE in 1h 24m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/34dea9a9210848dea20a2101b9a10b32 : FAILURE in 1h 17m 55s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91ad5247e1494e099cd860576d2285ae : FAILURE in 1h 27m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/249f7b23171c4505be0e0052f6f5b7d0 : FAILURE in 1h 20m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044af61a4dbc446080cfbb4e2a7723f3 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8f0c2d858d64a553f8366c604b2caf682a6e782b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 17:12:15.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"955a619877317596cf05e2226ffd9ad1725a434a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 17:12:47.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ac198b6c1cb659d3f6af7bb2a8316a2bbd91cf6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 19:52:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f84d9c2fc907413ebf21522ce4e0635a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38bd065dfd68410aa3fd1c8c4f4dcd27 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e850cced72854301a8a2e231b92fecfa : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4067c2e7131e406e8c2dfd38e57d1d5c : SUCCESS in 1h 54m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/66a50a23aa284d58802f21bcf27ba003 : FAILURE in 1h 14m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334310a01fc34977aa003c64503b03ad : SUCCESS in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb22aa8b3d02405fbf024f2d664849a8 : SUCCESS in 1h 48m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6d82b3e00d7402595f3ed184d6031c7 : FAILURE in 1h 20m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d134fcb1b24544f7aff21f1e4c075922 : SUCCESS in 1h 50m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/101fd4e198464ee1b1aa2d4fa6d69b99 : FAILURE in 1h 02m 18s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38525c69f9674faba5e60ce4234b7a34 : SUCCESS in 2h 36m 10s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/626bf831ae0a4f8b9cd83feee2542d45 : FAILURE in 47m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ec1cba09fa2409786f70b405fd5ba05 : SUCCESS in 1h 06m 35s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ddcf4f1398c42c88506f9ca56ac9196 : SUCCESS in 1h 47m 49s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae6d11317bcc48b983daac374cc2fb60 : FAILURE in 1h 14m 39s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/24fe2eed85734ffc961d47c50df33079 : SUCCESS in 1h 38m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe9624e9ae0b42a6b0462568896dfbc0 : SUCCESS in 1h 46m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6995d3cf6723420693d68d1f1d7fc101 : SUCCESS in 1h 30m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7925ffc705704e1ea39bbef5d888fcce : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f19a41f978b99e0fa18175f106f39b64b805a37f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 07:17:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85358a0f7e5be6de8ada3488fe6912720e7cedbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:24:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db8757ad40fe465eb7e6c5bffe457ba6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c5f683038ff459abc86bca4961ff97e : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f687e366fda744849f533f956cc320cf : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7938551fe1c243ffac35f840b1e64337 : SUCCESS in 1h 57m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/71bbfbcb1bfe4e6681c55dedd7138352 : SUCCESS in 2h 54m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/106c9a9d5d23415c86fdfb1aca6395d4 : SUCCESS in 1h 52m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fafd9ee4ebca48409ee0e377f6c06c44 : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d6e70ca00a94b7ea472e4aa68af494c : SUCCESS in 2h 03m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c6f51efe09f4d6ca123d7da2935bf18 : SUCCESS in 2h 01m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07984f0c69344e008997c748c52b7c5d : SUCCESS in 2h 55m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03c22830ed6743f8891e76d0e1d5170b : SUCCESS in 2h 41m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b94c521be11243df92e8733136fdbfb0 : SUCCESS in 1h 17m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e55b1348e6541379b5ea360ea55dedf : SUCCESS in 59m 10s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/1301bc3e15244acf9547cf972b0f770f : SUCCESS in 1h 58m 06s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6168e59c04884eebbc63e9ec76168e6d : SUCCESS in 1h 56m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5ae1824eaab1483b9f81848009dec640 : SUCCESS in 1h 51m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd54af611954c17ae2371ef9a57155f : SUCCESS in 1h 56m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83471784673d4e67b66e1e5def39ffb4 : SUCCESS in 1h 33m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961ce36bf3014d54b66bf9e40329ddda : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":4},{"id":"9111719f8094ac3de493a9a9f088b6807a2ad8fa","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-08-08 21:02:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3e194435bdb55248001aed62163fda3c2d145404","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-14 08:42:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"faf44d80260d9eb593e7e5a6ecd75a417d6f9e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 08:43:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54698c6fa5844c55b9ac72ad34f6bcef\n\n\nWarning:\n  Change 888517 in project openstack/openstack-ansible does not share a change queue with 888471 in project openstack/openstack-ansible-os_aodh","accounts_in_message":[],"_revision_number":4},{"id":"a8c1aae56e0a7cec6ab565a2dd949d6f7b7f5bfe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 08:13:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6fddc7c8e5f6e2430e176bc770397aa203ff4d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:00:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ff64b36a45341ec8ac78eca6e355ed7\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6325bbcad6434e3bb61dff17871cc962 : SUCCESS in 17m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a3144da5b6a24dd7a234830670eb6be3 : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cccbc68009f043b3bd6e7e1c6df66343 : SUCCESS in 1h 41m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73ce59f962a945fc84464b7e899575a2 : SUCCESS in 2h 08m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3090b4c21baf48068244fb493ac7bca7 : SUCCESS in 1h 42m 59s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8cb74e4d04844b196ccbd126fb927fb : SUCCESS in 2h 27m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1064c543b9634a2b9fa7408f38eb080d : SUCCESS in 2h 06m 26s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36be78a9f052476596b13e254f1c476d : FAILURE in 2h 43m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f006d758bc4e447b8a78d42b61f7a88c : FAILURE in 2h 28m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3dc984bd0d944c8a1b2221cc6397358 : SUCCESS in 1h 13m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db3192cd17874135966435608733f3a4 : SUCCESS in 1h 19m 17s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/65b22a3cb3704b43b07234522a4bf3d9 : SUCCESS in 1h 48m 06s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e121fb33caa148e9ac964206bb95cf64 : SUCCESS in 1h 55m 31s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2e054caf02fb4316bb18fbfaae22b5eb : SUCCESS in 1h 41m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91505b67ab2b46369be5f8562e7e09fa : SUCCESS in 1h 50m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27a3b4c0ca64ae4a54dbb92f550539c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"30ffa7b95c3f4673d7a41e7c84d29cc02343175c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 12:43:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca4e8ed6d9929e15bf81c798451a035b11662549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:02:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e80943bbb7d342a99b787f219ec489cf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b005fcdd3a7e4de0951e903fd2e89129 : SUCCESS in 13m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/94cbcf064a334f19935883298967034f : SUCCESS in 1h 55m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b241aa5d482d4a7e89deccd88f079d8e : SUCCESS in 2h 09m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d3bcc8a69b64da58d5a92bf160bd5d1 : SUCCESS in 2h 16m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8f40e4218924a31bfe52f461922f23d : SUCCESS in 2h 00m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/709f1947faaa4d7fba27201704f8e2e3 : SUCCESS in 2h 16m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/938131d3caee44e5ba868fad7f9c2b8e : SUCCESS in 1h 53m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74de41e678a44a4dae059138447e652a : TIMED_OUT in 3h 09m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe5153dd206a4bcba93675eca2287767 : TIMED_OUT in 3h 05m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f1acaaa616e4f118acfa8200a672734 : SUCCESS in 1h 17m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68db30c26752441d98b0f161aa20672a : SUCCESS in 1h 10m 08s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b04606ad561b400d990dce303a4ba53b : SUCCESS in 1h 48m 24s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48adaebe82e74190ae94cfd0fe4b580b : SUCCESS in 2h 03m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7a732de9a61748548d9605b3a7700a4d : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d8220be42d2403d9b9a29b32abc725a : SUCCESS in 1h 55m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42411a28370c4b3aa2ceea13da1657be : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":4},{"id":"94c193e5d79ade096cfc3119627b524bd289eb7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 07:31:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"04b686027932f5ef2c360aa01fdc076fbf3400af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:49:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58429a0378b94f60a81606591b251f55\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3123c7df924f43668e3d6bb1f2d2082e : SUCCESS in 13m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4643581b9e934791ab10f472aad3bf7d : SUCCESS in 1h 52m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0ebbc85f67648c48f39b3b8b10b08d4 : SUCCESS in 2h 01m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26715a66c72f4e0a92750571a4e8ee93 : SUCCESS in 2h 00m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5beec9ab9cd042508b6664a9d2bd7985 : SUCCESS in 1h 47m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/104f17fbc31a4addba86d31b66171ad5 : SUCCESS in 2h 11m 42s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a4d7f491770426191c56e97a70dce6f : SUCCESS in 1h 37m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26bf963cf34e4da9987915c77abbb80b : TIMED_OUT in 3h 05m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb4078fb22674984b52e78bda773f1d7 : SUCCESS in 2h 44m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9acb778b58544528e2ad2d12000ea86 : SUCCESS in 1h 11m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0acce3248a8d4eb09e4867bb1be0127c : SUCCESS in 1h 08m 02s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0694f12c8a584ca08f3be3e49903c3e2 : SUCCESS in 1h 53m 05s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0787571bb54e4015bb1ccda04c685735 : SUCCESS in 1h 58m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1a87549326c14815a3acc3fcac76f27b : SUCCESS in 1h 31m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d6b251d82a0407ca2008b20b746894e : SUCCESS in 1h 47m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0447abc2d654f14949f9eed175ef0e9 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":5},{"id":"1a47d504d2edf0a72e77b280abce513cb6c1feba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-28 11:23:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c564e41e22e318ef3a009e3e8ef473b8b87d8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:34:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75d579424b9142b48366e34b82e1591c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1fb4af498c94b36be40e974adbfb3dc : SUCCESS in 17m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/02d86ca9d9db4ae58ceac8325d815dae : SUCCESS in 1h 45m 29s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/28e48fd6cb6b4dd9a2015d42907b3728 : SUCCESS in 1h 53m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb4c24528b7447428367d09a6b564bfe : SUCCESS in 1h 45m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/969a68ddd3cb438ba3d0a9b734bdd76f : SUCCESS in 1h 47m 42s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b834b0c88a174b7eb1946899807a3bcc : SUCCESS in 2h 00m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d3838820b4e4b3980fa3dc9c74bd2e0 : SUCCESS in 1h 52m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8be25dede7164dbc8e3a182e931a6507 : TIMED_OUT in 3h 04m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cae8399557148cebbe5862bfe57eba9 : SUCCESS in 2h 37m 52s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3eb20102d78b4076aba97584c75cc818 : SUCCESS in 1h 12m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a6f704d870464aa4bedfdcf700b58c : SUCCESS in 1h 12m 55s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d2be1ac4a3d45639d783f5bb7032743 : SUCCESS in 1h 54m 10s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63b7271a1f6d414d9a938c510bbdfdf1 : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d03cd4d248594d7b961889c204e3393f : SUCCESS in 1h 35m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92641c3788484be2829275f2f0d447b3 : SUCCESS in 1h 36m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6707853316824d9288754f6e1d46b687 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":5},{"id":"b07a23e02855b1addbaedee9209d85f506fe28bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-31 10:03:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb4198d77ad15e8267598b46a1fd0f1afb77e93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 13:15:46.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/926419eeae2a4daaba75a378c0a50654\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e7ffc69f7084cf38cfb4b2504d6ec96 : SUCCESS in 12m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/88287a18a6d44586b106f92ae3823df2 : SUCCESS in 1h 43m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/39610d5075a2455697dfa4374fe84810 : SUCCESS in 2h 00m 12s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d7adc8135454487a0b877072daf353c : SUCCESS in 2h 03m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/231c6677f3774c09b7e260c7e307c0aa : SUCCESS in 1h 45m 31s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/555650c55d91422aa62d578714a80594 : SUCCESS in 2h 03m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb6dc0ad27fa4799855e4e79fe9f1dce : SUCCESS in 1h 29m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31bb4454535c461288a9fd7d43536c59 : TIMED_OUT in 3h 05m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b1b5113b6a44873a5fd1975a22618e0 : SUCCESS in 2h 53m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/21833aaf2998436eaa67f75055c5c290 : SUCCESS in 1h 09m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdd30ec1d0f141b8ba28a5eb92d75ff8 : SUCCESS in 1h 01m 38s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dbdec9764ac4a12ab27f8693f0dca58 : SUCCESS in 1h 53m 40s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac5dc061117d40cbb5a3ff5e12472f10 : SUCCESS in 1h 59m 44s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1809e2abee174b298aaf22487dfcd314 : SUCCESS in 1h 49m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/498bff58f730423f914f34e4b6205b58 : SUCCESS in 1h 44m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17f1ba00bee40db938fdd1fa9a7d6e6 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":5},{"id":"9f803b66adedee14585d661b4f738225a9854de9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-04 14:21:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e31f2539c17c5e7b529f666dd81db2b1e137d08a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 17:32:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da733d415f424ed89b5daedd46d2d226\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef885285947e4e87b1e7946f0f9ba5c3 : SUCCESS in 13m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ea1562456d0e4c8fa302909c2d334fef : SUCCESS in 1h 50m 20s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a24e29cc9ab84e1784825764c6009a84 : SUCCESS in 1h 52m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b2763b58037b4872a9f59e768599e122 : SUCCESS in 1h 57m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e19f2339ee4c618b2dcce3061a1d45 : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b7d9f91e7e64702a54f02d248cd575b : SUCCESS in 2h 15m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/798995412101486b96932c6f5a23a88e : SUCCESS in 1h 42m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/034e8ea9af5c461b93b01ee5f343589f : SUCCESS in 3h 04m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcf19c0aae0d455794abacf3db755367 : SUCCESS in 2h 50m 55s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6228fc08b8374f449f1d09586cd0410e : SUCCESS in 1h 02m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77bb9b22e43e40ebb562bf8d5a72f24a : SUCCESS in 1h 07m 49s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f44f28dfc8a24b599b19dcd28831df72 : SUCCESS in 2h 02m 56s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a7be5d70cd74def9b79fcafb2628233 : SUCCESS in 1h 49m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e3f9ef99bd744d9da556f4e958300e53 : SUCCESS in 1h 37m 04s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d52cdb91ac3e4210935c87a779d4b0c6 : SUCCESS in 1h 46m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9d9e7f4569482ea2a1924ef6bdeb78 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":5},{"id":"e907e8e0bc9e4904ae6080c238d741de534392fc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-09-05 07:28:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ea8020e1a6ea54d5ceb7d97f826d30670effb0fd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-05 22:30:44.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"77929c80a68bdc54aba0f5dc1c2325922b5d8c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 22:32:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bc2141dddc2074b5d33d4acd3726b41bc2b0be12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 00:37:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d448ee590094c37bed150cf2ddf5551\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/524db64ae3d44bc48065b5d41a076910 : SUCCESS in 10m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/91f5549c44144a60bb316939da8cb47b : SUCCESS in 1h 33m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1881b9e8b48b47f28590156e088d02d6 : SUCCESS in 1h 53m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e008ca2a52d4747b4bb2da051c3edcd : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/88ff091b345c4d7388e7d593a50251e3 : SUCCESS in 1h 57m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40b13e501b934b5e8b4cc21a417145b2 : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e10a6a0bca4b4800871aea2bf1e86ed8 : SUCCESS in 2h 00m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/85d13a90fd134a198037b8c680aea6af : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/379b38fb6d1648babf4f5f044292034f : SUCCESS in 1h 44m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72f17be221d40b3924c34171d3ad9ec : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"31a34ba80e24a2d8c66c5ab15e0bd5040c4f0291","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 00:37:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8f9f8bcf0ccdbd70a7a3c39ede66b25b5092b5db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 00:38:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83ca3dbeeae4496a9d56090ce9d30fc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417daec9093042fb808b1c4b7b7b6f9a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bf019ab6d79d441bf19885cd19f2d0402e9653a5","revisions":{"653e9ab00aff026e11c6bd2f9be18970813e7316":{"kind":"REWORK","_number":1,"created":"2023-07-13 14:11:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/888471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/71/888471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/1"}}},"commit":{"parents":[{"commit":"4b2134df19205a916e396bf0e1e18a9e263a78ef","subject":"Add TLS support to aodh backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4b2134df19205a916e396bf0e1e18a9e263a78ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: I53c0166beb446802641c701f6e43d5e317f9d54a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/653e9ab00aff026e11c6bd2f9be18970813e7316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/653e9ab00aff026e11c6bd2f9be18970813e7316"}]},"branch":"refs/heads/master"},"ca167f0a5f61253834f50198431962913f89502b":{"kind":"REWORK","_number":2,"created":"2023-07-14 05:36:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/888471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/71/888471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/2"}}},"commit":{"parents":[{"commit":"4b2134df19205a916e396bf0e1e18a9e263a78ef","subject":"Add TLS support to aodh backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4b2134df19205a916e396bf0e1e18a9e263a78ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-07-14 05:36:40.000000000","tz":0},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: I53c0166beb446802641c701f6e43d5e317f9d54a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ca167f0a5f61253834f50198431962913f89502b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/ca167f0a5f61253834f50198431962913f89502b"}]},"branch":"refs/heads/master"},"a7825870d158630e3f2c12d5e174e613dc6f2ec9":{"kind":"REWORK","_number":3,"created":"2023-07-14 17:12:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/888471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/71/888471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/3"}}},"commit":{"parents":[{"commit":"4b2134df19205a916e396bf0e1e18a9e263a78ef","subject":"Add TLS support to aodh backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4b2134df19205a916e396bf0e1e18a9e263a78ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-07-14 17:12:15.000000000","tz":0},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nChange-Id: I53c0166beb446802641c701f6e43d5e317f9d54a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a7825870d158630e3f2c12d5e174e613dc6f2ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/a7825870d158630e3f2c12d5e174e613dc6f2ec9"}]},"branch":"refs/heads/master"},"845e8369afaec36724accdcfd4dbc041a1f6b935":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-07-14 17:12:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/888471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/71/888471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/4"}}},"commit":{"parents":[{"commit":"4b2134df19205a916e396bf0e1e18a9e263a78ef","subject":"Add TLS support to aodh backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4b2134df19205a916e396bf0e1e18a9e263a78ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-07-14 17:12:47.000000000","tz":0},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888517\nChange-Id: I53c0166beb446802641c701f6e43d5e317f9d54a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/845e8369afaec36724accdcfd4dbc041a1f6b935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/845e8369afaec36724accdcfd4dbc041a1f6b935"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bf019ab6d79d441bf19885cd19f2d0402e9653a5":{"kind":"REWORK","_number":5,"created":"2023-08-23 07:31:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/888471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/71/888471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/71/888471/5"}}},"commit":{"parents":[{"commit":"eeb070c6823e1d25368d10a1d846972e30c99cb8","subject":"Merge \"Use proper galera port in configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/eeb070c6823e1d25368d10a1d846972e30c99cb8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-13 14:11:04.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-23 07:31:20.000000000","tz":120},"subject":"Fix linters and metadata","message":"Fix linters and metadata\n\nWith update of ansible-lint to version \u003e\u003d6.0.0 a lot of new\nlinters were added, that enabled by default. In order to comply\nwith linter rules we\u0027re applying changes to the role.\n\nWith that we also update metdata to reflect current state.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/888223\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888517\nChange-Id: I53c0166beb446802641c701f6e43d5e317f9d54a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bf019ab6d79d441bf19885cd19f2d0402e9653a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bf019ab6d79d441bf19885cd19f2d0402e9653a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
