)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~892355","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~stable%2F2023.1~I6299827be3828931f9c796edb655aae8f1fcff23","project":"openstack/openstack-ansible-os_aodh","branch":"stable/2023.1","topic":"cherrypick-osa/db_port-24yhx0eaik","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-19 11:43:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6299827be3828931f9c796edb655aae8f1fcff23","subject":"Use proper galera port in configuration","status":"MERGED","created":"2023-08-23 07:26:07.000000000","updated":"2023-10-19 11:45:25.000000000","submitted":"2023-10-19 11:43:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892355-cherrypick-osa/db_port-24yhx0eaik","cherry_pick_of_change":890093,"cherry_pick_of_patch_set":3,"meta_rev_id":"43555961c2282e6e9681ed01972f4a85bf540a59","_number":892355,"virtual_id_number":892355,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-19 11:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-10-19 09:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2023-10-18 03:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2023-10-19 09:42:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 09:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-18 03:05:44.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2023-10-19 09:42:26.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"e792559268da625a6809552644535dc284022969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 07:26:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"45ae56c1630a69fddab03070023e18c5b3f2d8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 09:41:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e10e8e180784833956bf3cb3117e6f0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/066eeaa45e5e4c40846754d5bb14927b : SUCCESS in 12m 15s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/04efaf9ba61f425c909468ad84b6fe84 : SUCCESS in 1h 41m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a0b70380df849afa122644cfeb9bd5b : SUCCESS in 1h 59m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c17afb619f204f4a8639e6252e92edd5 : SUCCESS in 1h 56m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3b6c9cc45f4c2ea6291d7251532b9b : SUCCESS in 1h 54m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33ab344996344dfcbe3d7f155ddd2199 : SUCCESS in 1h 44m 44s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6462515fe4cc46cf9177dfe27d4b16d7 : SUCCESS in 2h 00m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8711e7f5ead845a99466cc97c58e4cfb : SUCCESS in 1h 57m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1510bba603514cf2947a7b1f2fcab4b5 : FAILURE in 1h 09m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bd19f178cd64df0ab6975bebb48db44 : FAILURE in 26m 13s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c608ef60f704672ad1b52925f5000ab : FAILURE in 1h 05m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b5f36d2bd0114b45bc5eaec3c7b97a9c : SUCCESS in 1h 08m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ab0ced1e2b64b6ea5157a7f70ea43a1 : SUCCESS in 1h 01m 26s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/162d8168e40c490ebc94417eedf62959 : SUCCESS in 1h 43m 26s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf83dfcff61d4a60a0a1f24c3e974452 : SUCCESS in 1h 39m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/96010db864ed47d18492e7c1b96eaf80 : SUCCESS in 1h 38m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2033b6c82f214ff38bdd4f015404a466 : SUCCESS in 1h 52m 15s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c8b076515e64b95bacb0bce4827ceb1 : SUCCESS in 1h 49m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa77d8d781354944b62fefa547b076c0 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"457a7a0aaad915dc39e789b3031c247e5cb030b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-06 16:09:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47149fb6256023365bee88f3bd0ba9a76eeb46aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 18:51:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/811436aa4aa24098bb83e1a709ea2ea4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d54eb65bb541c8826f1197a3b8e227 : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/932418a9b02a4677bb2df2cdf7495343 : SUCCESS in 1h 51m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c24ac3c564da4e98af318a431baafc4d : SUCCESS in 1h 56m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eba13865ffb5453fb3ab3ab6c8dd7ba8 : SUCCESS in 1h 56m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de315483488f473eb0c910b38e6fad35 : SUCCESS in 1h 57m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d8bf2399756474ea06e39c7a7ce6dbe : SUCCESS in 1h 37m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/acbd56ecb7274dcf97a37dd3ac5e41df : SUCCESS in 1h 47m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14c8d695c36d406287dfdcd6be27ba30 : SUCCESS in 1h 57m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c00e2c4d474d4508a6552fed5a7f2830 : DISK_FULL in 2h 21m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/750bac481cff4b7caadf0320fbd5ead3 : SUCCESS in 2h 33m 25s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8008914137904f0085f8aa10fbe6b063 : FAILURE in 1h 06m 48s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79526107ac0d4070b9c6035d5baf027a : SUCCESS in 1h 13m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f8f4c9bac5a4713ad30c38446616970 : SUCCESS in 1h 06m 10s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1caff42d1834c4fba8644c279e7983a : SUCCESS in 1h 35m 22s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0e67c8064030460997d2d0af6df265a8 : SUCCESS in 2h 01m 35s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/63df6f2d197446ea93eb2a13380c1600 : SUCCESS in 1h 49m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10363f592b2e4b82b8ce5b21bbabed14 : SUCCESS in 1h 55m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6f7e1ab916348b6ac16d79c694e5edc : SUCCESS in 1h 57m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb31a3b533e4b3ea24a58da84688c23 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"168d3adc45240934fc97e432f38a2430f5d1835c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 10:17:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0601c137af386f8e67c5f252433469d075eab53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:03:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9c89e09764647a6b8651d6d54e7a573\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f8e911c55954fa6a22200f9dfd23ca7 : SUCCESS in 11m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a63159dee6a54c6c943f9cad048c7cbd : SUCCESS in 1h 46m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/72cf7339331340e5a9dcabc1b304573d : SUCCESS in 2h 01m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48db9eaaa06b46edad4910c45b8f8249 : SUCCESS in 1h 48m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5aa74467c9642d883ce10ee833743b9 : SUCCESS in 1h 57m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e76ef620461485fa91251748236db8c : SUCCESS in 2h 05m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb32ac1bb5f14c359a840eaec7ebc085 : SUCCESS in 2h 06m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c9ce1d62cd64bfdbe9698a54e2f833b : SUCCESS in 1h 41m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed8e3a4d75384214a8c52500e6a76b18 : DISK_FULL in 2h 40m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4422d98abc0d44d784ca86823bc899fa : SUCCESS in 2h 33m 49s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e6d10eee8e9429bbd2f1a9ab70ac2d6 : FAILURE in 1h 15m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/746722a8b62d4167b830afadb774d616 : SUCCESS in 1h 10m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91e7b99d9f554697b31e4ce65b961d60 : SUCCESS in 1h 08m 07s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7d90ae8b3e442eaa5f1d398d86a44fa : SUCCESS in 1h 58m 13s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9677353bd0de4641a24447691c0baadf : SUCCESS in 2h 02m 01s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0015bc07c3f24ce2829730c4bff52081 : SUCCESS in 1h 33m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/497e0669a4594b238a9e63855c25e973 : SUCCESS in 1h 54m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb82f3e6a18b412bb306370f6e0fe3a6 : SUCCESS in 1h 49m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4088c1a70046fcbf8626d6d666e288 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"82c75581042fba4cdca940ae9cc3dce972365603","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-10-18 03:05:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"591185a76f9c22a35bd11e4381fb4d016b117bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 05:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34535657f9e544fe84f7b23319a20cd0\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b01f6c440ebc4148858072ef5b98ff96 : SUCCESS in 15m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8df2f25020c94cae867e672e95dd495e : SUCCESS in 1h 16m 41s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/30b8caab7a7e45c69adf9abb08fc26c8 : SUCCESS in 2h 03m 11s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3838071e624d4e27a125cd5e4b3e0b52 : SUCCESS in 1h 59m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70a80384b7324c6490b5518be7e3b1df : SUCCESS in 1h 51m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fc5835f12864955a55f5e6134d0377c : SUCCESS in 1h 33m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfd163286681423387e4ef54f83f6bb0 : SUCCESS in 2h 10m 16s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a5028f0f5f14e639e62f8776ec19523 : SUCCESS in 1h 30m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4db324f0aa8e4ad7bd3914149ac5033a : SUCCESS in 2h 44m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/525bf097f3cc47abb1e5ba4ae30419dd : SUCCESS in 2h 25m 34s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/050070052cd444cda9ce5439fdc3b1aa : SUCCESS in 1h 51m 28s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0232dabf9c8048b1a8962c8f9aaec940 : SUCCESS in 1h 17m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c05920b2aef745dd86cdb77912fc2fa3 : SUCCESS in 1h 06m 41s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/45b423a87d334378bc77c009b7c6ffe0 : FAILURE in 1h 34m 00s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cdc0324800e0432f88cda9d5d8180572 : SUCCESS in 2h 05m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/29be448de63543e39398ddd5acae570b : SUCCESS in 1h 46m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e54d7dce26aa4600a357c7fb78d7847f : SUCCESS in 1h 46m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a0ad2e66e2840c5889b23d1923d44cd : SUCCESS in 1h 29m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4075329146142328ee3f12aaf3bba85 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"02f07e7c4cdb1713a79691398fb737e88afeeeb5","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-10-19 09:42:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f46cebe90946308d6e23dfd480017d28af3ca017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 09:44:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cce25c4b7a9dd9fc3448c481f18ec0f72485020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:43:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a2b662d5744d2bbbae84101f95e92d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03407645f5fc40de9c0b7b8b469af343 : SUCCESS in 10m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d3f0fc06fbdf417692aac5f98272e645 : SUCCESS in 1h 44m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec91b1aa95eb4b4985df6d89f0412eb3 : SUCCESS in 1h 32m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3edb309d49254e548d3a90a6de3b75b4 : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ffb7757234444171977e6f0118771359 : SUCCESS in 1h 54m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81c82c1ecd8e46779bc615245f5d7911 : SUCCESS in 1h 34m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e200a28717d472ebd737bf99665c91d : SUCCESS in 1h 35m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f3dcbd201aab4cb98f91077372ee1974 : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e150096409704185a794008b291fe1b0 : SUCCESS in 1h 27m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72fbb3371f9420d9b3fc5e872f42ce4 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"64819d4322efd7b5e7f5cf837d72de1438c891c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:43:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"43555961c2282e6e9681ed01972f4a85bf540a59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cde515df7ae4f2ca3dabbedcb82526d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40355e7b94a94f96b2c095fa22a41c64 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcbe61c54e1640155f149c1737be683be604958b","revisions":{"dcbe61c54e1640155f149c1737be683be604958b":{"kind":"REWORK","_number":1,"created":"2023-08-23 07:26:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/892355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/55/892355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/55/892355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/55/892355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/55/892355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/55/892355/1"}}},"commit":{"parents":[{"commit":"4a6a60bd522bbfa2d4b14df71396410f7ce5cf3d","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/4a6a60bd522bbfa2d4b14df71396410f7ce5cf3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 12:32:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-23 07:26:07.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888517\nChange-Id: I6299827be3828931f9c796edb655aae8f1fcff23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/dcbe61c54e1640155f149c1737be683be604958b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/dcbe61c54e1640155f149c1737be683be604958b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
