)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~917949","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I7229673bfff823467f7d756ef41713acdf49756f","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-14 15:48:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7229673bfff823467f7d756ef41713acdf49756f","subject":"Implement variables to address oslo.messaging improvements","status":"MERGED","created":"2024-05-02 12:13:11.000000000","updated":"2024-05-14 15:49:16.000000000","submitted":"2024-05-14 15:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917949-osa/messaging_improvements","meta_rev_id":"144ff5bf0dc345b079abae6fa9a6c7ee3ecf3d03","_number":917949,"virtual_id_number":917949,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 15:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-13 09:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 13:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-05-14 13:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 15:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 19:19:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-13 09:12:17.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-14 13:23:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"49f54c07d7b0f35a084daff68938fe9aeb866608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 12:13:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"946acbbc575073da6ba5a7bf6cbdcfe092172ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:20:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a91a2dd43d84a10825529b36088ffc2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/254885263a6247699e00f5f2bd25e1d7 : SUCCESS in 13m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8e5ab8c983a840319339f258df8354ab : FAILURE in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7dbad4de39be4563b55d79e038eb5db1 : FAILURE in 1h 11m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c28447cee5514efdb23586f973370d34 : FAILURE in 1h 12m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3533595b43184250aaae0edc9acf37be : FAILURE in 1h 20m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a04620c3fdd6483e8a858b2cc3c694c2 : FAILURE in 1h 10m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a652f087fb9946db9f17de6dc7dc9a49 : FAILURE in 1h 19m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b72dbee711145aa875f299b4362c570 : FAILURE in 1h 02m 25s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89221cb146d0454fa396dc02c1482a9a : FAILURE in 1h 06m 17s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f1c21deb7d43449c9975ba475a9e9015 : FAILURE in 2h 09m 20s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fbf48880a02c4ac79987ac7cba3dda4e : FAILURE in 2h 57m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43e58ed734dd4f11b096520aa2702855 : FAILURE in 1h 57m 01s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1beb662fad2e49f78e65010b6e8d6db0 : FAILURE in 2h 21m 26s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e54c352eee59405f8efb855ab883c99e : FAILURE in 2h 15m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58138e305afd4d15aad1272e5b08bcf1 : FAILURE in 1h 44m 45s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a95538a3ebcb4ccb87ac35c5539cd128 : FAILURE in 1h 03m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb5e9aa78ca74a4fa867bcf8ed8f5a2f : FAILURE in 53m 38s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9231a7fd1dc5403096d6bc94da235abd : FAILURE in 1h 42m 05s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87732eb28a4840c8adcf9e89d4488e3d : FAILURE in 1h 14m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/358364e5716b49f19528286b663e7497 : FAILURE in 1h 01m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ccf7178542d490a8759ff998765cee1 : FAILURE in 1h 09m 09s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eacac0adc644b6ca5f919bd70f6e68a : FAILURE in 1h 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79bcc694fe74c8bae9fb9c714d5c8fc : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1c0c0dc1b87dee2d6c20e57a0a2d6d92c3655356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-03 15:55:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"89b5423c0ed125fa1a249ab75bb7b50171371860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:15:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d8fe3f7dfc94f1a8e390324e6e25121\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe2094a5136b4cff874a9ea0fb584591 : SUCCESS in 14m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/029546c458104c55bea3f04bf083b0a2 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/019dc163371f484a89f95981dc44800c : SUCCESS in 1h 27m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6095bc0b6134eb7849b74bb1fa7ff9f : SUCCESS in 1h 37m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/306fbbbf130341ee87ed73aaa97af7e6 : SUCCESS in 1h 43m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e29f6e24ad284a38992becd694ba43b5 : SUCCESS in 1h 18m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfa05b6581514048949c390c883ce62e : SUCCESS in 1h 53m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3bd1f9d22e74914b75cf6ac17bcc849 : FAILURE in 1h 27m 58s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd41df51e2d456a8409e0a4f53401de : SUCCESS in 1h 14m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1cfa0bc60d4d4bd18535d9337ba409c9 : SUCCESS in 2h 23m 01s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55ac133364474620b7e12a8f58a34cd5 : TIMED_OUT in 3h 04m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7d872a36e57430db99cbd4898e52e76 : SUCCESS in 2h 21m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc263de8cb6449028ada2bfe43cb6681 : TIMED_OUT in 3h 04m 06s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b8caf130557407fa28a76fd93d4c197 : SUCCESS in 2h 40m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d00bd811288148c0b4ab7c153d150a9d : SUCCESS in 1h 50m 57s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/949c59f46ea44beeaa43705df8e6dbfc : SUCCESS in 1h 22m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e14585415f4352868cd7108f136967 : SUCCESS in 1h 10m 01s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f086f02742a74c42853a5416c2342c11 : SUCCESS in 1h 31m 55s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84ab7c89dbdb4edda09a921e18557faa : SUCCESS in 1h 35m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bcd1a9e3e6834673b27ed13790282782 : SUCCESS in 1h 23m 46s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0261ab4978514b02a381c5a6c6774172 : SUCCESS in 1h 13m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cdfd0664d2e499692a8ea305194e8d5 : SUCCESS in 1h 30m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb60eea3b3bd44c68237d069d2310186 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d54f2d56f0d068bc9f5d59889e58676e97e6b921","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 15:59:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4b05e3b15eb3c49c9be5a2de5a8152a0e9b10c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 19:11:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5de351dcdeb44b79bd7e7198fb5f9967\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9dfc68265d1493cad30b1820e9f7f4a : SUCCESS in 13m 42s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f06c1ab05c5e49dfa12a546c4e98c187 : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bf6c75e0834d4fd8a4f335efcb59fdcd : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b29a5f2d6ad34be8a0478f47ddb749f0 : SUCCESS in 1h 45m 38s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6d6410c35e314bc08cc5d974b024b879 : SUCCESS in 1h 39m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6f650fcc75b4f9a8b522e74e4c0036c : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6f1ca38436794e23833b9afdfbb2cf87 : SUCCESS in 1h 50m 19s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2de81bae38284a849fb416995c200d07 : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/650242f083304ec9894b5415c230423d : SUCCESS in 1h 38m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a5aca07c652b4de5828f84076e59537f : SUCCESS in 2h 29m 38s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b4c5eecbfd14897b1db237b62723f39 : TIMED_OUT in 3h 04m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b88157b771ec45eb9a5e1311f1b473a0 : SUCCESS in 2h 09m 40s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b85e7348bf314172858a731fa9ad77e2 : SUCCESS in 2h 47m 50s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2f0f6a179334397936815d9955d940a : SUCCESS in 2h 42m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf0ee7fa72e746dc823d821aee42f09b : SUCCESS in 1h 41m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df341614aa814dafb85645144daab246 : FAILURE in 25m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aadb2c5198c5438e8d3c5114224cb635 : FAILURE in 47m 44s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d3ab50c8bb949239ac849288ebb9c96 : SUCCESS in 1h 22m 21s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4a329657cee4c30bf2180a710d1d1c3 : SUCCESS in 1h 46m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9f305d14b38045bf94732e77650a4e7d : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9fa9810e065944439e30317b607648bc : SUCCESS in 1h 29m 27s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8502f85ec1e541988439b778210f7720 : SUCCESS in 1h 39m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2fcc85911147e2a49166d65bdd9200 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d57535e771350d5be82a71c5ff2c51e63f776993","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-10 19:19:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efa83bb74b31fea49c8119ce3e122eb29465b2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 22:29:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42a12189e0b84adfbb308822f81c3aa9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93730eed4f504b8fa685cc96b03f258e : SUCCESS in 14m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0282c5f71ca14495b6e11f75bd6c24d8 : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/439f2d45ce33458fbca1030a3d5b0160 : SUCCESS in 1h 33m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/87337b8470194582974603b572376b08 : SUCCESS in 1h 30m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c8a73ee050744e2b482235f67d33c30 : SUCCESS in 1h 54m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b4e4e3d259a4754a171e266bdedca31 : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2939bff414994f19b5235d173728d7dd : SUCCESS in 1h 35m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fc355922ff74ffe93e7631a735edef9 : SUCCESS in 1h 30m 55s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b04693ecee4448b2bc32d690b950be42 : SUCCESS in 1h 38m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4a6ade5e8f446f7a82eb717ad878984 : SUCCESS in 2h 30m 42s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6b0905ada374406ea50ac21e984ac736 : SUCCESS in 2h 58m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6aab67b6b5404434bd9b3198f4df917c : SUCCESS in 2h 25m 27s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88c5028cef0d4f9a91e4d00a44df2938 : SUCCESS in 2h 42m 02s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad098c5dcebe477d807e8a487b550f5b : SUCCESS in 2h 26m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/333f61585e6846f1a725b8636d6d9961 : SUCCESS in 2h 06m 02s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a2ecd652ea74b7dba7073ad5b64972c : FAILURE in 25m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5994763befe442794a28394ff2649cd : FAILURE in 53m 08s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9e5f62ae77b442bb463fb0dbf44122c : FAILURE in 23m 10s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c3d799f075f34cdbabb110797ab1da65 : SUCCESS in 1h 48m 02s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c0c14d8347aa4469ba5487df5930bc0f : SUCCESS in 1h 33m 23s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e0af5b75af34e209b44fcefc9bd94da : SUCCESS in 1h 34m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efbd3fec948a4a8c83ff841aa0c52c7c : SUCCESS in 1h 32m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b812b33060d43c596c77fed61e9323d : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"752bc6ea1f9e14bf9d31e6b846016c32a99ce7ab","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-13 09:12:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3568fad2f0c83877a6dc9fe70500889883da6f83","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-14 13:23:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c6a31b1f036a218a70737e203d7c2d8221e09030","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-14 13:40:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5285530f4eb89d7e35c644855a5bca2b3cf02df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:43:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c50dba7e42bc97405bf3078947a179513b7b93e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:48:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5705c7b7d2794f9dadec5388349d594b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9afaa181c4774a6a9b004e25f8e33802 : SUCCESS in 13m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e07220478db9414aa8e3ed42be31424a : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a2644e9b67ce440e83b8f66b5d48803e : SUCCESS in 1h 29m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47950f551e5346aa9940d177fbf5cf0e : SUCCESS in 1h 50m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df4bb5a1222c4d5ab26c705aff3bc000 : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fb9e65d080649f69ff693c5ad589c4e : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f723c02cec414359a16e12ad2840360b : SUCCESS in 1h 53m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e347a8011ac45908a9fb8a6c6b9f81a : SUCCESS in 1h 43m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5d6d80649a2747e28fdb51da21293583 : SUCCESS in 1h 53m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b1178016558c4c0bbee84d8fb2dbf5d4 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/df221dd38e9044b58bb95b06b4d0ddb1 : SUCCESS in 1h 37m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89b13298d8db4a9396529f5bddc00cb7 : SUCCESS in 1h 25m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15b489540264b438ba85efad3302528 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1cadba7a7e95b1d3d0737552718892f55228128b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:48:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"144ff5bf0dc345b079abae6fa9a6c7ee3ecf3d03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:49:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/090a6d37580e422faae8709987ba5422\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a359fa2206324f2ca4d9d30f86afb34f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"af6fe320eb2365a2c12e882dac4cb6bce227fee2","revisions":{"5540b4c6ed07c3d291f794a683cd7def86432443":{"kind":"REWORK","_number":1,"created":"2024-05-02 12:13:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/917949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/49/917949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/1"}}},"commit":{"parents":[{"commit":"2e63aa5aed6e39b59225a849ce34c51166b8f5f6","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/2e63aa5aed6e39b59225a849ce34c51166b8f5f6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 12:13:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 12:13:02.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I7229673bfff823467f7d756ef41713acdf49756f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5540b4c6ed07c3d291f794a683cd7def86432443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5540b4c6ed07c3d291f794a683cd7def86432443"}]},"branch":"refs/heads/master"},"88ea5d9ebb47553b29ed5569fb2a26c9cf8edb61":{"kind":"NO_CHANGE","_number":2,"created":"2024-05-03 15:55:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/917949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/49/917949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/2"}}},"commit":{"parents":[{"commit":"f694169e6f22f7868c3a05be1846e7403956ad87","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/f694169e6f22f7868c3a05be1846e7403956ad87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 12:13:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-03 15:55:30.000000000","tz":0},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I7229673bfff823467f7d756ef41713acdf49756f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/88ea5d9ebb47553b29ed5569fb2a26c9cf8edb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/88ea5d9ebb47553b29ed5569fb2a26c9cf8edb61"}]},"branch":"refs/heads/master","description":"Rebase"},"af6fe320eb2365a2c12e882dac4cb6bce227fee2":{"kind":"REWORK","_number":3,"created":"2024-05-10 15:59:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/917949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/49/917949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/49/917949/3"}}},"commit":{"parents":[{"commit":"f694169e6f22f7868c3a05be1846e7403956ad87","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/f694169e6f22f7868c3a05be1846e7403956ad87"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 12:13:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 15:59:21.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: I7229673bfff823467f7d756ef41713acdf49756f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/af6fe320eb2365a2c12e882dac4cb6bce227fee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/af6fe320eb2365a2c12e882dac4cb6bce227fee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
