)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~941320","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I9677f44a6b801094070850c103543ffa6cb60e0f","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa/fqcn","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-04-10 14:02:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9677f44a6b801094070850c103543ffa6cb60e0f","subject":"Auto-fix usage of modules via FQCN","status":"MERGED","created":"2025-02-12 07:44:12.000000000","updated":"2025-04-10 14:02:57.000000000","submitted":"2025-04-10 14:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941320-osa/fqcn","meta_rev_id":"811a97277c58632e0e0cbc13f520a89e6d462c3d","_number":941320,"virtual_id_number":941320,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-10 14:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2025-04-10 12:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-03 20:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2025-04-10 12:06:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 10:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 20:56:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-04-10 12:06:04.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"e1d837e14529f1cf3b4c28079c6890c685301052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 07:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c58ce521e2c3178ae1e6deb8faf347af5aff9a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 10:24:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/897b0aa2b0174496971ec4697fa586eb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbad1cff569f40d294ca841059d72099 : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5cc95ed7a4604da9aedc3864985ed21c : SUCCESS in 1h 32m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/58f98992552c4ac9b95baf99a48916e6 : SUCCESS in 1h 37m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d767d7c62a6441b79e75f81781b858d8 : SUCCESS in 1h 40m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db536c408b814879b0f626cdd2c8e90c : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ebe19c3d1924795be2eb1de9a4d6d42 : SUCCESS in 54m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8605fa940f984f0b97c6178b3a9b9366 : SUCCESS in 1h 38m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a306959eed5a40279c6e15398d7f37ed : SUCCESS in 1h 36m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbed6b58341c44ac97f15431f60ed849 : SUCCESS in 2h 05m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51dfa6fd31a74cc7ad5e6a58f9150221 : SUCCESS in 2h 10m 38s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a01a100bcd7a496985ccd444ba693429 : SUCCESS in 2h 03m 45s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac95f51ecfbd4801a15657de80bbe57d : SUCCESS in 2h 00m 27s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1f9ed1a180a467bbc75e10fe44b7a23 : SUCCESS in 2h 01m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/118190bdeca44fd0a11478b665e12bc6 : FAILURE in 2h 08m 59s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b01fd4f59d544a22ad4cec93ca91b6ec : SUCCESS in 2h 18m 40s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df776715bb894be486bd9491b7692b1a : SUCCESS in 1h 07m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef27ad3f27e644f9923fe1350f964662 : SUCCESS in 1h 06m 16s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/668920de78344bf19b8792f621872861 : SUCCESS in 1h 32m 26s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a7ed1e734b8a49e7b7b6c6efb4dc1b63 : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3fc6390da0946118743aeea202bec6c : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc10b6a70fe7479a975162dccffc4f69 : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16e9ceaccaa34c0fa81fc1f37b17adda : SUCCESS in 1h 29m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d10bfc51078463a8551115ad3a0929e : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"556dbcabde2bd3b3ce3791277cb5fa0fcdcbd6c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-04-02 12:30:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b62a562369c628c872218ee9af11752e276a61dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 15:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c19e10a08ec64060a27efcf45efc6228\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8b19e09f59c4c86a548e42f74533a73 : SUCCESS in 7m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/021479ef5c4148969a99f85f13b8f234 : SUCCESS in 1h 20m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e69ca3d6d99446b0ad9ce24218566e61 : SUCCESS in 1h 18m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/134265f8d52d40f4b6bafae25f890fd4 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/074f225e854a4ed9bf4252493115ed00 : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6668a09aafea400fbc802f6ed1898609 : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dee958ed7bdf4f199653668b1ed5f378 : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b77713f2b67941b7b2a3a98f46e7b9a6 : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4545f46a4a8f41bc92f710305589e013 : SUCCESS in 1h 43m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/add11fa8d1524107b7991c12194af423 : SUCCESS in 2h 10m 04s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/69c77ff34be2418cab3d9c8b75c56813 : SUCCESS in 2h 04m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecca63682c06453d867935f88dd34727 : SUCCESS in 1h 03m 19s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2e1affeaf3340888bf09b7f8d171294 : SUCCESS in 1h 51m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f0d520c0e81a4d56bf847623ba798f0a : SUCCESS in 2h 22m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd2f49e138d6467a8872cf39bd90f057 : SUCCESS in 1h 00m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06bb67cd57ea4b5ea84d5734a7c776ea : FAILURE in 57m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4301f7169794b40afd877596bd1bb08 : FAILURE in 1h 03m 13s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd2b37dabbc44e1780f14f6fb9e0f7bb : SUCCESS in 1h 37m 02s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/37d866a5ec9e4cb093f5aee4663da3d6 : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/19ec28be7ca54af9a2639561f6e42c30 : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebec181e1b094bc9a3ee022ed7ab8a63 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b216527e6cd491cb15aec9f53787e33 : SUCCESS in 1h 40m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad53bccd93a47a298b9b043c1cc0390 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"82424e8be54f50dc30f7041deb0cb460ed2c467e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-04-03 20:56:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"871dac6671011dde2eb10ac03a6418ff447e2b71","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2025-04-10 12:06:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de29d4a23e57f60b3d8032f712a1cae194f0e6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 12:08:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d7497e158d5008a86c60601fd437c07cd66a4d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 14:02:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47f1903eccb40e89ed09e24264f9c79\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51a30714d784063ae41446919f4d8a3 : SUCCESS in 5m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8685a331c05f4321aec930f7896ffe48 : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4253c44b1d184b478de20b1283c97406 : SUCCESS in 1h 39m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/677705c94624412e90c277d300e3bdf7 : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b57bbb7ef9904df8babd240032523749 : SUCCESS in 1h 34m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f51f401433364f86857a1da972d015aa : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f8b98979951426792010f8e7878bab7 : SUCCESS in 42m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f6e69e8c50f4cd68040fef68b86fa65 : SUCCESS in 1h 42m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/41322edb214341f7ad0e00f688c055c5 : SUCCESS in 1h 30m 25s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/706d1027c3a04e5785f39daaf763b483 : SUCCESS in 1h 14m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5888a40946f4e7a867693d244fd2e38 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9a2939a4dfe4cc7b7249e79f50f2560 : SUCCESS in 1h 48m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3fb4d3028d4e0390deeed658ef6d53 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3646ff4c8c660733ecb1e695965a4c6ba1babd6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 14:02:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"811a97277c58632e0e0cbc13f520a89e6d462c3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 14:02:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e0a5dc7f0474f67a5d3f7c0c0227f01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b387f4dd76409fb3d86b92dfd203e8 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb06623482d4d4ba638a30dd0ce46de9bb927234","revisions":{"bb06623482d4d4ba638a30dd0ce46de9bb927234":{"kind":"REWORK","_number":1,"created":"2025-02-12 07:44:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/20/941320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/20/941320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/20/941320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/20/941320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/20/941320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/20/941320/1"}}},"commit":{"parents":[{"commit":"08e44039e805bba73192ef832a805ea1e866482c","subject":"Auto-fix yaml rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/08e44039e805bba73192ef832a805ea1e866482c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 07:44:01.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 07:44:01.000000000","tz":60},"subject":"Auto-fix usage of modules via FQCN","message":"Auto-fix usage of modules via FQCN\n\nSince ansible-core 2.10 it is recommended to use modules via FQCN\nIn order to align with recommendation, we perform migration\nby applying suggestions made by `ansible-lint --fix\u003dfqcn`\n\nChange-Id: I9677f44a6b801094070850c103543ffa6cb60e0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bb06623482d4d4ba638a30dd0ce46de9bb927234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/bb06623482d4d4ba638a30dd0ce46de9bb927234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
