)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~956290","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I9b5f7350f5a567e0c8f89d5a3847f450be17e830","project":"openstack/openstack-ansible-os_aodh","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-09-02 10:01:13.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-09-04 19:38:29.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-09-04 06:10:32.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I9b5f7350f5a567e0c8f89d5a3847f450be17e830","subject":"Remove outdate file manual-test.rc","status":"MERGED","created":"2025-07-31 18:46:47.000000000","updated":"2025-09-04 23:55:59.000000000","submitted":"2025-09-04 23:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956290","meta_rev_id":"43be2b3031686eb0707a3067aac944eb587a27b9","_number":956290,"virtual_id_number":956290,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 23:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-09-02 10:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-04 06:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-04 06:10:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-31 18:47:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-07-31 18:47:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-07-31 21:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de52062b6f7c4b4d1a7d5e3b4db7524cca07a663","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-31 18:46:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e7507eb40f0106d18b6e307f634f5c3388789a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-31 18:47:29.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"6508cf2c68925050cee2532e840e6af4a680f541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-31 18:53:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"be423435ea8bf092974ba746732e28d3204d7d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 21:20:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/860e4bb17d3145df8c498e1009821c32\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7852e020d8b4490ab5da8d3f128dab0 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/250285b0233345b29fb8feca3fc4419e : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20ebc05abb90443f93184df6e6ff6ca7 : SUCCESS in 48m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edfcbedb28cf4750bd2ef0fc86ef4ec2 : SUCCESS in 1h 14m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8da754ff897144f5b14ddbf194dba2bc : SUCCESS in 2h 20m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73447b2f90f64bc984cfac6eb031651a : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef32e8113d414502be306f35ed4d0e97 : FAILURE in 35m 22s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/fda96d6f725f4ae6a588dbb04465d8de : SUCCESS in 1h 39m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cbcd4b5335b412d9923252e3a5a80fe : SUCCESS in 1h 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a3b1a0002a4163a14f59778c78f2c4 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"62c6c64445ebbc9cd263ee3bc10bec5f5b644392","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-02 10:01:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"42a49c896e8d88ba2ebe129fedfa5a630234e66b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-09-04 06:10:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38dc47d1363a08750e8c24dfb8c71fb4ba16e59a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 06:12:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a34a39accdea8bef3bfa83816c87c6504363f526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 08:03:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ca5ae0ef96c46599ec7b80e2a2cea92\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23c60d6d65ce49fc8953a407a57b0e3a : SUCCESS in 6m 15s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b6d5c288515a40ecb7e1e37c1e65c036 : SUCCESS in 45m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c76dd5cf929d4d75848dc6fc75d12ec3 : SUCCESS in 1h 44m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/049d1c8baeee4da39e0fe8456f3461bf : SUCCESS in 46m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e8a3e807fc64de799de1116da395d43 : FAILURE in 1h 22m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e42f4a270e964bd5817099c26b857718 : SUCCESS in 43m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3353f60fea054eee9916d13e9e59cf85 : FAILURE in 1h 21m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30f335d92ec44991ab0e2ddc3240e84c : SUCCESS in 1h 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3acf9c8aa392471a97a62c386d6bca97 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"5e4dd66872bb619ac3c5df07c433962bdffa3e5d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-04 19:38:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f956b677de7b2067daac66db3d3a410e27884acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 22:04:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c54767a20cf432295e2161edab004bc\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a583279d42124f6baf9bb3b2836fc8ba : SUCCESS in 4m 02s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0c6b7278cf8f400b98b51a958887d9e5 : SUCCESS in 1h 30m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0332c78849354af891475ff6f0e4ccb7 : SUCCESS in 45m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5da1fa751ac4481ea6e1aa65605f1d83 : SUCCESS in 1h 50m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f122015b5644c5caabf92bcabbedee1 : SUCCESS in 1h 39m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7872394239c40e29cebc99a32f629d5 : SUCCESS in 2h 19m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a624a1df97e47ef87e8066fa182ef56 : SUCCESS in 2h 19m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8658760f906941b5acea4cba08eda0ae : FAILURE in 25m 29s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4e005afd30554423931646ba76d1bc44 : SUCCESS in 1h 29m 18s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e2754a10cf24a2aa944cb8817c7cb73 : SUCCESS in 45m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09dba8eb45bb480991734ad6c527f7d2 : SUCCESS in 1h 23m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf6c6fc5b1f49328869afd4ec2c2ebd : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"583fbdedb5f53ce760a1dd8206881fad7bbfdbc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 22:06:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"416dd025a720f323a7ae385a22a53cc371a0d584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 23:54:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0062437390ec4d9bb4ad7553ea2f9b47\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a43d212b4504dd2b2dc5ff22fa4b8b2 : SUCCESS in 6m 17s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1d6b369b97a04f358e47dae86e24eee8 : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/311daf9dd8f14ac3a2fd8e86b277a085 : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ead59951d844fe88c37265663658846 : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd778d62d2664d16b7a4da19b2101edc : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/314f0dad4e9a401f9ee3af14d37813c9 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8079ca55bf244c05873f38257116a974 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74d482cbd8374c48afa48797551b3b82 : SUCCESS in 56m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edaa4ac863049fabad718fff4e91865 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8b2bd4ec1ec51fbc7f388f518c43611f47034e1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 23:54:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"43be2b3031686eb0707a3067aac944eb587a27b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 23:55:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb5221d68f84b49b0f0eec0b958939c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade9a53db4ce48efb015373bda9da3ab : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"04e2966a3e212239f74a74f2712055f93072440f","revisions":{"759c94a7cebb0474534bc9801d8918415beddaee":{"kind":"REWORK","_number":1,"created":"2025-07-31 18:46:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/956290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/956290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/1"}}},"commit":{"parents":[{"commit":"d77e7c1a466109bbb992fbd6db5a19d9c6e442d5","subject":"Drop unused [api] port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d77e7c1a466109bbb992fbd6db5a19d9c6e442d5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-31 18:46:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-31 18:46:47.000000000","tz":0},"subject":"Remove outdate file manual-test.rc","message":"Remove outdate file manual-test.rc\n\nChange-Id: I9b5f7350f5a567e0c8f89d5a3847f450be17e830\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/759c94a7cebb0474534bc9801d8918415beddaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/759c94a7cebb0474534bc9801d8918415beddaee"}]},"branch":"refs/heads/master"},"04e2966a3e212239f74a74f2712055f93072440f":{"kind":"REWORK","_number":2,"created":"2025-07-31 18:53:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/956290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/90/956290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/90/956290/2"}}},"commit":{"parents":[{"commit":"d77e7c1a466109bbb992fbd6db5a19d9c6e442d5","subject":"Drop unused [api] port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/d77e7c1a466109bbb992fbd6db5a19d9c6e442d5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-31 18:46:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-31 18:47:22.000000000","tz":0},"subject":"Remove outdate file manual-test.rc","message":"Remove outdate file manual-test.rc\n\nChange-Id: I9b5f7350f5a567e0c8f89d5a3847f450be17e830\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/04e2966a3e212239f74a74f2712055f93072440f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/04e2966a3e212239f74a74f2712055f93072440f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
