)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~981599","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~Id2ff2fb57c25ff75f70f83c7c030faaa382febd9","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-20 12:52:01.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-20 09:21:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-19 11:03:24.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-20 12:52:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2ff2fb57c25ff75f70f83c7c030faaa382febd9","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2026-03-20 20:24:08.000000000","updated":"2026-04-20 12:53:46.000000000","submitted":"2026-04-20 12:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981599-osa_hashi_vault","meta_rev_id":"b873763ec210a13788d46e0bfd9303162fe6f152","_number":981599,"virtual_id_number":981599,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 12:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-20 10:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-04-20 09:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-20 10:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 22:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-23 14:18:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-15 20:40:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-15 20:40:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-15 20:40:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-20 10:37:04.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"34ed291b72a88d242d6976eff0ab40efefaf4f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-20 20:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e53599070c82540b422e63b7d138ac06816334f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-20 20:30:40.000000000","message":"Topic set to osa_hashi_vault","accounts_in_message":[],"_revision_number":1},{"id":"a955f47887fac7058b2b56622d22604b6fa018bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:35:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d882e1304974f58b3b918e077578c79\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7536b9d7c5dc43f393b6a3040f558f7c : FAILURE in 14m 10s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d67d3ab36e264881a7e3fdf8eb58bc58 : FAILURE in 37m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9abf79d96f2c461f92c6fadea2394041 : FAILURE in 30m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/52419606683e49eaa1964e572f3d8736 : FAILURE in 33m 00s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/092ca9c39788469fb203f2c5f3769fa4 : FAILURE in 34m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/84cf2686686c4ac5bfb0601127227803 : FAILURE in 22m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72683bceae2d4d67a004f1b9c9b1a9e9 : FAILURE in 33m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57a31638c9c4426c8254ca91dbc3c830 : FAILURE in 36m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9514c034c7a445cafe645edf45b37f9 : FAILURE in 51m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6e5c27a0cd9347ca87925d894a4c5bed : FAILURE in 51m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d355fd393594909baf59ff99ed55a0b : FAILURE in 1h 37m 14s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbfb6f7afc414503853a5afe56100501 : FAILURE in 1h 38m 46s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f85f69d90bb499d811afcdd25d152e3 : FAILURE in 1h 48m 35s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53385a0c7b33426db89f1c8c68c7e197 : FAILURE in 13m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0513d8c9cb848d8b9691a604843ada2 : FAILURE in 26m 23s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/94855956dcec4a719aa9ba3def5777d3 : FAILURE in 39m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/60f79136a66b46a9b74597f92d70c3b3 : FAILURE in 33m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/74abc1e1f4e14ff799248920a097ac87 : FAILURE in 33m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7160a3149de843e785cfafc8cebf2c05 : FAILURE in 29m 41s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/438adba8dbf54e6d908936a2377be359 : FAILURE in 34m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1519e059fbf4d9993b148ff70311b73 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f0a99fcc4601bf2492d78dcbaca5dfb9525971cd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-23 14:18:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"845996b12507f77871f4e377e1be876f0423817c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 16:08:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce1f3eb7659141df96e38b277758c33b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8ea680ecc634b3aa5130c439c19311d : FAILURE in 12m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/7e1bef0f208c4d05a414e1bfefbf3463 : FAILURE in 27m 01s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c144783c80ce40f597f8a669b0824bfb : FAILURE in 31m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/392a450632794921b4320c5df4cd2277 : FAILURE in 27m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb5eee71ec3445aa903ba9a78f1f0b06 : FAILURE in 36m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3c1d925f4cc5432d89b9af910ed2107a : FAILURE in 31m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b2a938a4e2145d69ae66fcd463620f0 : FAILURE in 31m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51015de872d341129090a9a604e1f143 : FAILURE in 31m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b8ddab373524436b24284b9e3b9ecd8 : FAILURE in 1h 43m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86749df605f04b0982e06a9908ea674f : FAILURE in 50m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5487e398fefc4d898abe492863e7ec4c : FAILURE in 50m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26b4f5c18efc4ea295dce44b928fbb8e : FAILURE in 1h 43m 18s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23df1c8309a74881b6ecb60881e6631b : FAILURE in 53m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c4ee2f1d3f78455d87335354613693fb : FAILURE in 13m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/287db2de5ba34f69b9544e65f98ed1ed : FAILURE in 23m 12s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f452a5338c4944d4bcb1174c44a42121 : FAILURE in 25m 58s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8e75ce25e44d4e379696a5b37f83c4e4 : FAILURE in 31m 52s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a59c0cba3a134f02b1d672ee4282b1f6 : FAILURE in 26m 38s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/37b8e94aee7a40c2a851a26ca0e2387d : FAILURE in 32m 22s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2957e26c4d945df80a7e6a2704c7981 : FAILURE in 20m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ff6a743bc74559a493826b5b502f3b : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"aee55f5370407d2dfe80935c2535f8b3b432ab12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-24 12:10:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e10865a6fd006347abdd4d2f59a5fa5dd83092c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 14:39:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffb0a420519e4f009f079f42f9936e25\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d32c145eb894eb79e10861dc88fc938 : POST_FAILURE in 6m 14s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8e4b2ca1f39347acb8178a49cdf03acf : FAILURE in 1h 29m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/908946a3e43a4bc5b76dec940c558802 : FAILURE in 47m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/06b4ce3c8b8140178cd20b81827608fb : FAILURE in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/64d13dfc2d934bcab465b76348ba3671 : FAILURE in 1h 37m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a8996f1363d1489fab18c376675de308 : POST_FAILURE in 1h 10m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/043580164de24377804520ddf78f3a68 : FAILURE in 1h 29m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d5ce4cdfed648d9875c45814f1dc9ea : FAILURE in 48m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f7e7a43db0c94f84afc7ceb2afec0805 : FAILURE in 2h 15m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/41c20e2afe6c41c1ba923a0c98eabc3c : POST_FAILURE in 2h 01m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2a1baaa23cf4d0ab8540b255b3b8a37 : POST_FAILURE in 1h 49m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbc73fb4c35c450a99d0d06b87e91795 : FAILURE in 2h 22m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0829c1322eeb420b96e6841d3cd5bb87 : FAILURE in 2h 21m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/523577f0893842bd938f693d464d9fcb : FAILURE in 8m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6339a5b6cc5a436e8abc94a3b619c61c : POST_FAILURE in 1h 04m 45s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/606b8305d0c748f5a5af05ae35a08dcc : FAILURE in 48m 24s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3630c31b217c4f9482796c3c772f7965 : FAILURE in 1h 05m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cd00e0ac1f6044eaab2fd3bf53f557d7 : FAILURE in 1h 24m 36s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b9279c7e4908447fa781e8b982f183d3 : FAILURE in 1h 30m 51s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2343ad103c1b42a0833d19b4807f64b6 : FAILURE in 1h 27m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683ffb1da85643eca5a5c8281eedca95 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ec2e4b10b728ff1e621f2deb2f0234814f0202ce","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-31 11:00:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63baa9a0d6ba6a42e91670022afffaf51fd432ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:28:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c079c7e056b48e58dda93439df6bac2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7731aa8167f2492d92d04323c547c141 : SUCCESS in 12m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/912785b6548e41d59566620983e4c5fb : FAILURE in 1h 24m 43s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9e27142609cb436cacff0b62e45849e2 : FAILURE in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/42d6cc139e884003a63c3135fba510f1 : FAILURE in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f61f62a8e58140fb8cf3b70392e2dd01 : FAILURE in 54m 55s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f096014a1494ef7940cf1f475e380aa : FAILURE in 56m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e2b47413df742b1bf03406af7a847e5 : FAILURE in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434b4fdea53e43cdacb2a010427ba87b : FAILURE in 1h 40m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3f4e01642201441a923cff15529a6400 : FAILURE in 2h 08m 59s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6afbaf05725644a2b324e9b7b6be255e : FAILURE in 1h 19m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef8063f60dbf4a9db2f9d596e75248a0 : FAILURE in 2h 09m 51s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9e9755f70924bdc8efbcffcb79cefd2 : FAILURE in 2h 10m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e939be04a2814528ab59901447873fb8 : FAILURE in 2h 20m 06s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/25352f4d1a044767af2eade604f7fe9b : FAILURE in 10m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ac4c2b5387454794d47e292bcc0c65 : FAILURE in 58m 20s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b3378ce1163f4efdad0ca01710a586eb : FAILURE in 1h 22m 48s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2530fa0ae7244a93b826ea60ae968ef5 : FAILURE in 1h 23m 32s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ff40cbb62f784044803ca49741fa1aba : FAILURE in 1h 23m 00s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2437e77acf8e42d59c87d600ae62e3e2 : FAILURE in 1h 23m 08s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72386d52501342cdbe62f3c66392d127 : FAILURE in 1h 28m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b29fcba54243a7a8cab9201483f452 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b12572ca1249c7d58a75ee334ac554fcdfef7f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-15 09:02:03.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ae3444063d0b72da627224493e78105c9bba0da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 12:22:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4c59dd1b9794dd5bf6237cf24424c88\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1dea06e266f4c90a92d6a9e91aca5ba : SUCCESS in 16m 57s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0362ff8aaa3d43ce997204ab8b387cf1 : SUCCESS in 1h 27m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f74d08111ef949d4979a3083c1a77e8a : SUCCESS in 1h 48m 19s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/50a7a4a7fd564b4c86c91e60d6d410b7 : SUCCESS in 1h 36m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8ff0bc0c014425f9370c1a755f44c67 : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/85af2dcb9c89416cb84566e0c14f7899 : SUCCESS in 1h 42m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5b87df4616948bcb41d08685b736415 : SUCCESS in 1h 43m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912d58250f6946f3a5018670435456cf : SUCCESS in 1h 35m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b3be7e20ffdc497e8072340f8d4510d4 : SUCCESS in 2h 49m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b67c2cc5c984cf58a56696825293fb1 : SUCCESS in 2h 47m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3c5e26001684dbeb6e1a5190e3887dd : SUCCESS in 2h 47m 30s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad1a4cedf74e4e229557bd1128fb49d0 : TIMED_OUT in 3h 11m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17cffdbd5b024e6e98c627f051c0d6c8 : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/21eff5f923624c1095db1a47ab1051e9 : FAILURE in 12m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/699b7fad68aa475ebfaa1ffb25512a4b : FAILURE in 42m 51s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/626fa13f47384d8a9683864e0016ad1f : SUCCESS in 1h 44m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7fff37a687e541078fdd064410274f9f : SUCCESS in 1h 45m 47s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bb5b7a907f854673a2de9fbcf4b03315 : SUCCESS in 1h 45m 10s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/540bb4979c8f44bab3b930427f6f7348 : SUCCESS in 1h 00m 54s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2a469ff4faa4b55be2e849057d0dcc0 : SUCCESS in 1h 41m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d790027e2864541a2872ef7ddf5aeb4 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":3},{"id":"b241d9d2f32351efbbb832718f7008aee9c3b11b","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-15 12:45:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10aafe2b68eb5f7a49264f7a5ac83cb89808ce18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:52:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a2725afcb049bc8a665703cfcc90df\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0254bcb7ffee4d00b9352403b6f9332a : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/83bebef1cf77421c934635a40e8f8e78 : SUCCESS in 1h 43m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/20b9fd99f34e42abbb49a0616c5689e3 : SUCCESS in 1h 01m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c697297043d24e3988dabb12c526860e : SUCCESS in 1h 06m 08s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4bdcb2693be340d8b6b6a6b24144b501 : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d1ecfbe0fe64d66bf3a3c5b63134248 : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75ea20a0e4244b6390e4018131d64877 : SUCCESS in 2h 05m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4894388e22f47468cbed53d38d8a356 : SUCCESS in 1h 11m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/328471c07d554cd1ac650061fb700255 : SUCCESS in 1h 24m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/961a5719ab304b1c812e024cc4672b4c : SUCCESS in 1h 26m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b7597ee08b943768d9ad9d65f2c7a98 : SUCCESS in 1h 24m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aefe394f84bd4dfa962b019672fa655f : SUCCESS in 2h 34m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21e0c09450034e58915cab0a93c91d18 : SUCCESS in 2h 55m 26s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/05cc79fc74f041a69e72f7e10f920ea4 : FAILURE in 13m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f2c04c9f3144ec29164c7b1f54a3968 : FAILURE in 44m 58s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e28756f31edf486cb15d428dbc07d8f4 : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2db9c8b6f40b442d955e3545e3e66fdf : SUCCESS in 1h 35m 13s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3deffe1df31b4279b4f6abb67dbf671b : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/10d27ae5e1854433a6f0d6717b10d917 : SUCCESS in 58m 21s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9263a58a62e44606b38a75a51f8733fd : SUCCESS in 1h 31m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5d792211764ab3892de26e0aff1ac6 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":3},{"id":"de10960f671f5f56eea99d33209b3b65c7aaf62c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 20:40:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5afef1aff322bfa8630a433f2a2b4da1c69f14cc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-18 08:09:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b26c9a9a42c19237b383243b20bd622a653ca295","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-18 08:12:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cddd3d67791c75adc9382845c85b8733dd5f2f6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-18 10:47:28.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c5ac67bbdbe874115db8119c10b0012c9174ada9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 14:16:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b406d98d7874bef90e362b3e20ddadd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a63abbbc6f841ab82d80251187ebd10 : POST_FAILURE in 12m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/349755cab9df45d48cfe023378cf57ea : POST_FAILURE in 1h 18m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e705d31e99e5482fa645b67b0e73f5e3 : POST_FAILURE in 54m 18s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f9c3c383f3ea4b4b95315f8a6dc2cdd1 : POST_FAILURE in 1h 42m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b6722fa2e7e3431cb555c4891713eb28 : POST_FAILURE in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ca16d03fe89c4ac48adfbd4ae6392b2e : POST_FAILURE in 1h 02m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/963316a26aa04d9685ccf719c6e7730e : POST_FAILURE in 1h 41m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63cf1f2149254a899e403a54380b6572 : POST_FAILURE in 1h 52m 03s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f71c8f01270a4508b38fea63bb0773fb : POST_FAILURE in 1h 26m 48s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a87a6a6f0fa745039f318254019a11f5 : POST_FAILURE in 2h 44m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07e194642ba6408d8e6da7e39a370380 : POST_FAILURE in 3h 11m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bc3c766a8854208919fc95f4c29141c : POST_FAILURE in 3h 10m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de6ff77c5736476ca4abb54cdfb1b9d2 : POST_FAILURE in 2h 56m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae28621594fb47c6b08c2bb4788b73b1 : POST_FAILURE in 11m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b0f717222024c32a0c3e557760ef7c7 : POST_FAILURE in 1h 14m 58s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c0f4c530c71c410a8ef6741b81ad06a5 : POST_FAILURE in 1h 02m 30s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8d548a6b0ead4843aeaf51a600077922 : POST_FAILURE in 2h 04m 16s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bdb147742e1840999e06dd7b374c777f : POST_FAILURE in 1h 40m 49s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/58016fb56f5d42e1a18b8889086e7be9 : POST_FAILURE in 1h 24m 12s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5901945a458b425db8c3c16c2dfb72e4 : POST_FAILURE in 1h 41m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5dd666f4a743499ae8526ca46ed5ce : POST_FAILURE in 6m 33s","accounts_in_message":[],"_revision_number":4},{"id":"90ddcf3740b07af5e5658f024472434ca740306e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-19 11:03:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"deff75d274f0568901e213fd0ddf248940f10d88","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-19 11:03:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f36d2335b6ff1ca2bc08f2c29ef9fa554b928420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-19 13:45:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01eb38b34a9f48f790b17c3559b0972e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85a502277ae34b34b0c7ddbffc8346bc : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/6ca712a7916843f383291df0a0165446 : SUCCESS in 53m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4a0877350d534aafa82722d61a8cc2c7 : SUCCESS in 56m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a8947c1c88f3479984401585912cc36f : SUCCESS in 1h 36m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24b9fe1342954ba8a2c94b3469686cbb : SUCCESS in 54m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f7c9d2e84b4c4548aaf9a1d1fd972676 : SUCCESS in 1h 50m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4bade33ed7249ce9d6d376b8958ff02 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c3adde2b0464af28e1696ac879b0b25 : SUCCESS in 1h 42m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc0809233037454897cadd6d59e4c728 : SUCCESS in 1h 21m 17s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eda0d8aed029427bafedb4acb2ccf5d6 : SUCCESS in 1h 15m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e8452a0db4d414b8262c3d7710e82e3 : SUCCESS in 2h 31m 35s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67695b0f15fb4bd9add0cf52e1465235 : SUCCESS in 1h 12m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/733b8a76cede41d8aa68205e32e09cca : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3af3fec865364857bfe9efcf2363d352 : FAILURE in 9m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd093e29e4694d839627e282e673af38 : FAILURE in 34m 02s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f76962de2f494c2fb2349388ffef7162 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/73bc9bbf9fec4babbceaf68e5802cf52 : SUCCESS in 1h 39m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0165142fd2544110a65391602ade54ce : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2a87fcaa3f30470fb88c2d578afa4aa7 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/530c9b37f2f44ec5830f967e39899c07 : SUCCESS in 1h 40m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f30f14b25154346965280611da4d80b : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":4},{"id":"e81c24585531451be2c093fe26abe8f356f46ea6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 09:21:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c2e1f1aee6f6dcecca3172e4f870d5f00f2525b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-20 10:37:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a07a5371f50b72188ea7f04203b13a3254fad707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:39:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d9ddd31ddad28c96aad8623c25e93444fd6bc92e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:52:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668d55bf80e4422da654f1e77e8edbfe\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b198576304c43c78d1cbcdb531886f8 : SUCCESS in 13m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/295991c52e5641c288979a8d7f15fa96 : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a2c04df595c4693b1448468034c30ee : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/de7fe9b6c9214395b650e63b23cff405 : SUCCESS in 1h 43m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f9c7630d4dd34cff94593c4c9d57c455 : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/26a5d83c5d284b6d921505a39dbdadd2 : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6437d4bb2890488c91931f5de196237a : SUCCESS in 1h 54m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11060efa9c294f38b9254cfad097d396 : SUCCESS in 1h 58m 33s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a2ddf39a52794b1f9dbd6ad9196d1de4 : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3eb3ce309f42464ab66bbe2b134849d4 : SUCCESS in 1h 49m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/bbb82eb5ecaf4acf99d20598a068a3f6 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b001b23e645b4409bffd8a4ed80441f0 : SUCCESS in 1h 39m 11s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a07bc164c85743078c99f8d7436fc624 : SUCCESS in 1h 47m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4600b8ca0cfc4181b355a3df91923f40 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a0807aa7ef8e6279a56002d0a294f6a93a9cb9cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b873763ec210a13788d46e0bfd9303162fe6f152","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:53:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baf157c990944b43950773c9387161cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d07e72debf241c5b899501549a699c0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9b04bea37f54b16c7febf7b2059496fc0aee0deb","revisions":{"0db632eeaf4d15f7c1ad6edda90497ae1204c335":{"kind":"REWORK","_number":1,"created":"2026-03-20 20:24:08.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/99/981599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/99/981599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/1"}}},"commit":{"parents":[{"commit":"8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:23:51.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:24:06.000000000","tz":60},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: Id2ff2fb57c25ff75f70f83c7c030faaa382febd9\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0db632eeaf4d15f7c1ad6edda90497ae1204c335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/0db632eeaf4d15f7c1ad6edda90497ae1204c335"}]},"branch":"refs/heads/master"},"7c5d4f5ec585899ce0a33043a0ffae7a602c0b63":{"kind":"REWORK","_number":2,"created":"2026-03-24 12:10:22.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/99/981599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/99/981599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/2"}}},"commit":{"parents":[{"commit":"8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:23:51.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-24 12:10:12.000000000","tz":60},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: Id2ff2fb57c25ff75f70f83c7c030faaa382febd9\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7c5d4f5ec585899ce0a33043a0ffae7a602c0b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/7c5d4f5ec585899ce0a33043a0ffae7a602c0b63"}]},"branch":"refs/heads/master"},"cdaf8ea76f481e21dea32043d0b016863ae661af":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-15 09:02:03.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/99/981599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/99/981599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/3"}}},"commit":{"parents":[{"commit":"8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:23:51.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-15 09:02:03.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_gnocchi/+/979877\n\nChange-Id: Id2ff2fb57c25ff75f70f83c7c030faaa382febd9\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cdaf8ea76f481e21dea32043d0b016863ae661af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/cdaf8ea76f481e21dea32043d0b016863ae661af"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9b04bea37f54b16c7febf7b2059496fc0aee0deb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-04-18 10:47:28.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/99/981599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/99/981599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/99/981599/4"}}},"commit":{"parents":[{"commit":"8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/8d2ce782b5f6af4db58c1c2a0851ab0f1e9af423"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:23:51.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-18 10:47:28.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `aodh_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_gnocchi/+/979877\n\nChange-Id: Id2ff2fb57c25ff75f70f83c7c030faaa382febd9\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/9b04bea37f54b16c7febf7b2059496fc0aee0deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/9b04bea37f54b16c7febf7b2059496fc0aee0deb"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
