)]}'
{"id":"openstack%2Fopenstack-ansible-os_aodh~985572","triplet_id":"openstack%2Fopenstack-ansible-os_aodh~master~I783900c2d740c53fe1bc84ebbdac1d4e5e51ecb4","project":"openstack/openstack-ansible-os_aodh","branch":"master","topic":"openstack/openstack-ansible-sync/sync-files","hashtags":[],"change_id":"I783900c2d740c53fe1bc84ebbdac1d4e5e51ecb4","subject":"Updated from OpenStack Ansible Integrated repository","status":"MERGED","created":"2026-04-20 23:14:44.000000000","updated":"2026-04-21 10:47:35.000000000","submitted":"2026-04-21 10:46:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985572-openstack/openstack-ansible-sync/sync-files","meta_rev_id":"643b2bd969c25dd0a3d1d43be400418e9082d316","_number":985572,"virtual_id_number":985572,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 10:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-21 08:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-21 08:14:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 03:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 08:14:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"2eac558e9d0d406a6bfa8ddcd414f7712a9f3fc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 23:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bbc8340a3049a0d0a38cfb88f74b4400fd5e511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 03:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ad4a1b3b9242fdb5e956397e783c87\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82aeb63880f949fa97714399155b7817 : SUCCESS in 11m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/38a4d83818a84bc58dcae4be80370c3c : SUCCESS in 1h 27m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4f72898106774600902c807e8361857a : SUCCESS in 1h 41m 33s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c5877e5fc4964984ae4d420e0832b777 : SUCCESS in 1h 32m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0625b22ac3bc41f58ce5c2c747ff97f3 : SUCCESS in 1h 02m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/08a9c8e3cd894ed1936dccb832b0f944 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de0b9cce6c4c44b98190a5e9f2bf049e : SUCCESS in 1h 44m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a1bd05e7f83489ba01cb8b8bb288f25 : SUCCESS in 1h 32m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/16e141e9adc6474698fe4c7bdcee957a : SUCCESS in 1h 24m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/600aae9b7ad74099a5cc01ab55196531 : SUCCESS in 2h 46m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb92ca15ec75471ca7c065b3823c66ca : SUCCESS in 2h 51m 36s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb241e50aab740f395b97d94ad22c999 : SUCCESS in 2h 19m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/209e206ede044e33ba1a0a948367c0d7 : SUCCESS in 3h 05m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6d813cff08549c59c48c075664f8c5e : FAILURE in 11m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20fd5172b64744e5be7f98696d916a42 : FAILURE in 1h 20m 18s (non-voting)\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/cbcb354ffa48458cbb63cb75d322ca7d : SUCCESS in 59m 14s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8877c96d0c6b4010aa0b07f985ef30d8 : SUCCESS in 1h 37m 05s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/010c96fff2704f8cbbbe39f070eb9b56 : SUCCESS in 1h 30m 06s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/230a48d291694a2094b2bc522f05ec69 : SUCCESS in 1h 03m 56s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a13cd4656cbb45be95bdd8f2e9e1b184 : SUCCESS in 1h 41m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797e7b1fa93b4433adc1225b889fb1d6 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b033c99f9c42d5b580153a1807a82d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0682ffb715bc44a5b88ad032bccdb5573995054d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 08:14:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b641998ae362581e4061d6960129a36e9181078","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:21:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1fa3d0261b512c0b53879dbf26b0202d2728ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:46:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7f2ae3599042c280d44172791f4b0d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23f6d998f6e4f3f9ee692a1a1634c27 : SUCCESS in 13m 05s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c0454fa655d3438fbe63c4d79aca90d4 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/02273f803d4f479f908a70161d82e2c0 : SUCCESS in 1h 43m 08s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/1275a3b1fc47481aaca3a8eec12069f4 : SUCCESS in 1h 38m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e36553b5656a49f48cfb18954e9c6021 : SUCCESS in 1h 51m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f11e1631f3914ea5a980e214b89a639d : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7fb3111d9154925a2d5440e5209ea50 : SUCCESS in 1h 43m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78df018053a9487cab02c864c4da86bb : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_telemetry_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/bc4d5778bd8d4535bf8f9caa3ac23ac3 : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6d715a38a56e4b789b61fca7253c5d4e : SUCCESS in 1h 38m 57s\n- openstack-ansible-deploy-aio_telemetry_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0410902af71543b7aa3c9721dee742ef : SUCCESS in 1h 34m 03s\n- openstack-ansible-deploy-aio_telemetry_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a5fb49bf6aa74aa593cafcd385bfaafc : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_telemetry_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db2201e3d9d8428db633df769254ecc3 : SUCCESS in 1h 07m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b1de162a004e72918892f7308b1feb : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/421b4901bda942349e1099584d51b3e8 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"091556db7e8c3e026f842700d0cad40082c6b889","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:46:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"643b2bd969c25dd0a3d1d43be400418e9082d316","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 10:47:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4665f212b4d94e1db9b34278f3c2ebe2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30aeb06e6b544108af51fbd44bce643 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5910d60cebc4b8993931c52c61a5caca68486d70","revisions":{"5910d60cebc4b8993931c52c61a5caca68486d70":{"kind":"REWORK","_number":1,"created":"2026-04-20 23:14:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/985572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_aodh","ref":"refs/changes/72/985572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/72/985572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/72/985572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/72/985572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_aodh refs/changes/72/985572/1"}}},"commit":{"parents":[{"commit":"92fd2fb013d46c38c1f283cef129f65204ff0f59","subject":"Merge \"Avoid leaking internal url for authentication error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/92fd2fb013d46c38c1f283cef129f65204ff0f59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: I783900c2d740c53fe1bc84ebbdac1d4e5e51ecb4\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5910d60cebc4b8993931c52c61a5caca68486d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_aodh/commit/5910d60cebc4b8993931c52c61a5caca68486d70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
