)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~658396","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~I7502be20bec4243f607087b6f74d20b1ff11a4d5","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I7502be20bec4243f607087b6f74d20b1ff11a4d5","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-10 15:27:30.000000000","updated":"2019-05-16 01:47:50.000000000","submitted":"2019-05-16 01:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658396-1557971270374-b2ea1c39","meta_rev_id":"119cbc78b2aa9a562c5224dafc5934185d2a8e8a","_number":658396,"virtual_id_number":658396,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-16 01:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":2,"date":"2019-05-16 00:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 21:19:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":1,"date":"2019-05-16 00:08:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 21:19:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-16 00:08:12.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-16 01:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b8557bbd33c431f56711b846039473cc172dd50","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-10 15:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"567d6323d9a0d0d93a47122ec34089768ec9a475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:01:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal-debian-stable/deaf0fd/ : FAILURE in 17m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal-centos-7/2ecbb85/ : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/5a18d10/ : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/50b527e/ : FAILURE in 15m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2194122/ : FAILURE in 37m 57s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/96/658396/1/check/openstack-ansible-linters/ea48bbf/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/96/658396/1/check/openstack-tox-docs/73dda16/html/ : SUCCESS in 3m 33s\n- openstack-ansible-deploy-aio_metal_barbican-centos-7 http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal_barbican-centos-7/4f83caf/ : SUCCESS in 1h 17m 43s\n- openstack-ansible-deploy-aio_metal_barbican-debian-stable http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal_barbican-debian-stable/f438be0/ : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic http://logs.openstack.org/96/658396/1/check/openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic/e11e87e/ : FAILURE in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cd2c8350025db81ab45179b3b2654c302507937d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 10:08:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a20fa7ec4c74852b23fa996920e47eaaa823bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:36:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/658396/2/check/openstack-ansible-deploy-aio_metal-debian-stable/1b0c8c3/ : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/658396/2/check/openstack-ansible-deploy-aio_metal-centos-7/c0eef91/ : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/96/658396/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/cc47f30/ : SUCCESS in 1h 05m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/658396/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/154f5b7/ : SUCCESS in 56m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/658396/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e5bcd9d/ : FAILURE in 45m 17s (non-voting)\n- openstack-ansible-linters http://logs.openstack.org/96/658396/2/check/openstack-ansible-linters/7cb1944/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/96/658396/2/check/openstack-tox-docs/bc3aa61/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"567bf37b7db85d7afd652f72cef3f96ff395d461","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-14 21:19:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2b50459ba7aa21f302cc8690745afb6272e219c6","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-16 00:08:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a0fed8b612c38a108b58257080328d971f0950ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 00:08:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"791d466528924b76cbf7a86d20e80e6b9d6b1307","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:47:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/96/658396/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/3ac3ee6/ : SUCCESS in 57m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/96/658396/2/gate/openstack-ansible-deploy-aio_metal-centos-7/df478d7/ : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/96/658396/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/78b08c0/ : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/96/658396/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6ecf722/ : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/96/658396/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/07d314a/ : SUCCESS in 59m 32s\n- openstack-ansible-linters http://logs.openstack.org/96/658396/2/gate/openstack-ansible-linters/7f877a8/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/96/658396/2/gate/openstack-tox-docs/bc94194/html/ : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d21686044d275e341c4a58cd955a64e8d818dbf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:47:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5590ef6a81b21d38dc1226a985e3c198863602f4","revisions":{"7026ab65dfe0d569136149e83ae9624cf910ef02":{"kind":"REWORK","_number":1,"created":"2019-05-10 15:27:30.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/96/658396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/96/658396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/1"}}},"commit":{"parents":[{"commit":"d3fbd1289fd6393a0b8908a57b17508ff5ff12ac","subject":"Merge \"Add bugs url link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d3fbd1289fd6393a0b8908a57b17508ff5ff12ac"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:26:25.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:27:14.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWhile this role isn\u0027t immediately exercised (the playbook\nis skipped), it is good enough to add coverage of the templates,\nshould we adapt the template in a global fashion.\n\nAdditionally, it is a problem to have different jobs\nfor SUSE support, as we want to use the integrated jobs template\nas a mean to get consistent across roles, so that\nit\u0027s easier to monitor jobs.\n\nChange-Id: I7502be20bec4243f607087b6f74d20b1ff11a4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/7026ab65dfe0d569136149e83ae9624cf910ef02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/7026ab65dfe0d569136149e83ae9624cf910ef02"}]},"branch":"refs/heads/master"},"5590ef6a81b21d38dc1226a985e3c198863602f4":{"kind":"REWORK","_number":2,"created":"2019-05-14 10:08:37.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/96/658396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/96/658396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/96/658396/2"}}},"commit":{"parents":[{"commit":"d3fbd1289fd6393a0b8908a57b17508ff5ff12ac","subject":"Merge \"Add bugs url link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d3fbd1289fd6393a0b8908a57b17508ff5ff12ac"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:26:25.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 10:07:37.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nChange-Id: I7502be20bec4243f607087b6f74d20b1ff11a4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/5590ef6a81b21d38dc1226a985e3c198863602f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/5590ef6a81b21d38dc1226a985e3c198863602f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
