)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~658406","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~I9b8bbec7d5aee3a385a947f0a88b53ba725da089","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"dynamic-aio-scenario","hashtags":[],"change_id":"I9b8bbec7d5aee3a385a947f0a88b53ba725da089","subject":"Use the dynamic aio scenario","status":"MERGED","created":"2019-05-10 16:02:20.000000000","updated":"2019-05-16 17:41:35.000000000","submitted":"2019-05-16 17:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658406-1558028495853-38c90c1d","meta_rev_id":"bf98d118b012581ffe92aa6863689d6a97275d18","_number":658406,"virtual_id_number":658406,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-10 18:41:56.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-16 17:41:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-15 23:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 21:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2019-05-15 23:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 21:23:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-15 23:41:38.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-16 17:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec7794c941f36571cfc8601d7444b34ed476942","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af1b249adfc11c5953c0a9e87aa9f07678a1ed15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:05:04.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f9b67a5ab4650a128f1a8bd1da04d764e752d975","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:05:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a4536fffa7e14093755084e077b02584f950057a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:14:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/658406/1/check/openstack-ansible-linters/b8efbce/ : SUCCESS in 4m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-debian-stable/b156e8c/ : FAILURE in 2m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-centos-7/5c978e6/ : FAILURE in 2m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/f9100f3/ : FAILURE in 1m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/925260e/ : FAILURE in 1m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5bea856/ : FAILURE in 2m 11s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/06/658406/1/check/openstack-tox-docs/c8f5e59/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"df307f2ff3865df034b07afb2a88352b1206b4a9","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:19:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"099c3c6d9c3d38179f110c5fbb64e8009b7877a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:24:17.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"bd63ac7be5956c83ab489637b827bc9ac880e8ad","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:27:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74e5d960085d9ed639b32a4ea9320770da10fe9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:39:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/658406/1/check/openstack-ansible-linters/27d5e7d/ : SUCCESS in 4m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-debian-stable/01c1f10/ : FAILURE in 2m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-centos-7/cf2bf53/ : FAILURE in 2m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/7110d04/ : FAILURE in 3m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f315e94/ : FAILURE in 2m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/357bb3e/ : FAILURE in 1m 45s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/06/658406/1/check/openstack-tox-docs/f40d735/html/ : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1},{"id":"565cae15cdcb78c4af38788f6b11383dcb78b533","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 16:45:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d350886abd597b2547abe51f371cb7209655026d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:58:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/658406/1/check/openstack-ansible-linters/764675e/ : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7328b8f/ : FAILURE in 1m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-centos-7/2f9c1ca/ : FAILURE in 1m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/7955193/ : FAILURE in 1m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b293a9b/ : FAILURE in 1m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0a011b0/ : FAILURE in 1m 50s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/06/658406/1/check/openstack-tox-docs/30c55f8/html/ : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9322f43059e4ae60e89908da3deac2dda1ff3bfb","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 17:19:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92a8281f615f45b9a5c527d65480af423542e2e3","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 17:42:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f0d32deeff5bf6921d76e3af4dab7e7f2a3680fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:02:15.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"0918e3f37e53db5f99809e068ead01703cbde9a3","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:04:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"34914da3237a627540fc87c1f53a4b7fe7b3df21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:07:33.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f1770c6be1b2476d4a8bbf0ca5fa1c313342c116","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:10:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ee7ddd1993fc467500c35b1cca264d82609d347b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:14:58.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a07da6e6f50317821cb0c007917e0787bcc2215e","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:19:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"087d468f460ef65c51120c2a277c66568443eb19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:39:22.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"57ab90f6d21a1c228f8a37d20c5892d7e08a9f97","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 18:41:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"452be6dcfaf4c04d357207afb1ce3d22b17411da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:13:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/658406/2/check/openstack-ansible-linters/028d20e/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-debian-stable/d3e7127/ : SUCCESS in 58m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-centos-7/df230fd/ : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/5ccdfb4/ : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/276d3e8/ : SUCCESS in 57m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/dd997e7/ : FAILURE in 45m 14s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/06/658406/2/check/openstack-tox-docs/52f1b9b/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ceae54ffd7505237a24ae078c09fc2704becae45","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-15 21:23:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82e121512e71bb464fd06789f9f6b60e324dfd12","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-15 23:41:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"979e94012cf2ae924fa932316449ecf1e4540533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:41:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d0ba07aec83b241843ee318736beaa319c0ad78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:07:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/658406/2/gate/openstack-ansible-linters/358cb33/ : SUCCESS in 5m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/bf554a2/ : FAILURE in 24m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-centos-7/997f0cf/ : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/0ce793d/ : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/cde9731/ : FAILURE in 22m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/4a218e6/ : SUCCESS in 51m 24s\n- openstack-tox-docs http://logs.openstack.org/06/658406/2/gate/openstack-tox-docs/296d604/html/ : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d965077bd313f745376f1f308554a1e5fc29dbb4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-16 13:19:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"634d0ef127b562855f52613349c5a1efbe200156","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:51:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-debian-stable/4f6f0fa/ : SUCCESS in 51m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-centos-7/3689d60/ : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/cb3ae43/ : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/000356e/ : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c056be6/ : SUCCESS in 1h 00m 28s\n- openstack-ansible-linters http://logs.openstack.org/06/658406/2/check/openstack-ansible-linters/72adbb7/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/06/658406/2/check/openstack-tox-docs/36f8a2e/html/ : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d1d905015843e2536533a83bf50b5d5aa755b6aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:52:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e7e8adc3ce694c3b092d2d991008754d916a72e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 16:30:05.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"600102afaeda49a61f6c4b52d6730477a03fbfe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:41:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/f2d3fa8/ : SUCCESS in 59m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-centos-7/6a726b1/ : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/1a39193/ : SUCCESS in 58m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c8cbb1e/ : SUCCESS in 58m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/658406/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/c69b90e/ : SUCCESS in 58m 02s\n- openstack-ansible-linters http://logs.openstack.org/06/658406/2/gate/openstack-ansible-linters/d69b3e2/ : SUCCESS in 8m 24s\n- openstack-tox-docs http://logs.openstack.org/06/658406/2/gate/openstack-tox-docs/892ccdd/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"46dd5ac5b1723d3f0f56b33371f0f07e96e151ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 17:41:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2693e9dffd178640fe46d8f25f23892fafff4611","revisions":{"c4a72c3e8cdac453ac34c6cff95102ec79721488":{"kind":"REWORK","_number":1,"created":"2019-05-10 16:02:20.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/06/658406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/06/658406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/1"}}},"commit":{"parents":[{"commit":"d5510747ff0c6fd58ca88ebbfadba7a1e98d4105","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d5510747ff0c6fd58ca88ebbfadba7a1e98d4105"}]}],"author":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 16:02:06.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 16:02:11.000000000","tz":60},"subject":"[TEST] Use the dynamic aio scenario","message":"[TEST] Use the dynamic aio scenario\n\nDepends-On: https://review.opendev.org/658398\nChange-Id: I9b8bbec7d5aee3a385a947f0a88b53ba725da089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/c4a72c3e8cdac453ac34c6cff95102ec79721488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/c4a72c3e8cdac453ac34c6cff95102ec79721488"}]},"branch":"refs/heads/master"},"2693e9dffd178640fe46d8f25f23892fafff4611":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-10 17:42:03.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/06/658406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/06/658406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/06/658406/2"}}},"commit":{"parents":[{"commit":"d5510747ff0c6fd58ca88ebbfadba7a1e98d4105","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d5510747ff0c6fd58ca88ebbfadba7a1e98d4105"}]}],"author":{"name":"Jesse Pretorius","email":"jesse@odyssey4.me","date":"2019-05-10 16:02:06.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse@odyssey4.me","date":"2019-05-10 17:42:00.000000000","tz":0},"subject":"Use the dynamic aio scenario","message":"Use the dynamic aio scenario\n\nRather than have to define a job per non-aio service, we have\nzuul introspect the origin test repo and add the service name\nto the scenario variable on the fly. This allows us to simplify\nour job definitions and simply use the AIO job template for all\nrepositories.\n\nDepends-On: https://review.opendev.org/658398\nChange-Id: I9b8bbec7d5aee3a385a947f0a88b53ba725da089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/2693e9dffd178640fe46d8f25f23892fafff4611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/2693e9dffd178640fe46d8f25f23892fafff4611"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
