)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~658587","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~stable%2Fstein~I05a86ab9e9e3ebde608604b876f004626c03d8f6","project":"openstack/openstack-ansible-os_barbican","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I05a86ab9e9e3ebde608604b876f004626c03d8f6","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:58:16.000000000","updated":"2019-05-12 03:33:06.000000000","submitted":"2019-05-12 03:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658587-1557631986495-72f24b91","meta_rev_id":"5e3c62a5e1e3f472664e1e357e0a51d9bef549ea","_number":658587,"virtual_id_number":658587,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:55:35.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20c7aa86a79e15717698d326eda5057bc216087a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:58:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33a3d806595bbdd0a11c499df71bd3de012300f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:24:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/658587/1/check/openstack-ansible-linters/e242501/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/87/658587/1/check/openstack-tox-docs/24f24c6/html/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes http://logs.openstack.org/87/658587/1/check/build-openstack-releasenotes/6541fe3/html/ : SUCCESS in 2m 12s\n- openstack-ansible-deploy-aio_metal_barbican-centos-7 http://logs.openstack.org/87/658587/1/check/openstack-ansible-deploy-aio_metal_barbican-centos-7/261e5a2/ : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal_barbican-debian-stable http://logs.openstack.org/87/658587/1/check/openstack-ansible-deploy-aio_metal_barbican-debian-stable/4a475e6/ : SUCCESS in 48m 38s\n- openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic http://logs.openstack.org/87/658587/1/check/openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic/74a0286/ : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"025d578b353c79290996f89c2f57e24a50deeb3e","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:55:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41fcfb4c58eb655f01a76f015836416878dad52d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:55:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c6418128b155252483e56915af17e142130f1f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:33:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"42bc86477edebbe84e82b17fa9ea56e2f12618d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:33:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/658587/1/gate/openstack-ansible-linters/770f1c8/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/87/658587/1/gate/openstack-tox-docs/5c7a4fe/html/ : SUCCESS in 2m 47s\n- build-openstack-releasenotes http://logs.openstack.org/87/658587/1/gate/build-openstack-releasenotes/f66645d/html/ : SUCCESS in 2m 27s\n- openstack-ansible-deploy-aio_metal_barbican-centos-7 http://logs.openstack.org/87/658587/1/gate/openstack-ansible-deploy-aio_metal_barbican-centos-7/a7ae59e/ : SUCCESS in 1h 25m 00s\n- openstack-ansible-deploy-aio_metal_barbican-debian-stable http://logs.openstack.org/87/658587/1/gate/openstack-ansible-deploy-aio_metal_barbican-debian-stable/ccddde3/ : SUCCESS in 46m 13s\n- openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic http://logs.openstack.org/87/658587/1/gate/openstack-ansible-deploy-aio_metal_barbican-ubuntu-bionic/5e8a993/ : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff5207066cfef8eb380c35d398a884d357e2738d","revisions":{"ff5207066cfef8eb380c35d398a884d357e2738d":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:58:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/658587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/87/658587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/87/658587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/87/658587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/87/658587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/87/658587/1"}}},"commit":{"parents":[{"commit":"06b3338c6572fe23a3bf8f84bd0a9e051fa5e3f5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/06b3338c6572fe23a3bf8f84bd0a9e051fa5e3f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:15.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I05a86ab9e9e3ebde608604b876f004626c03d8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/ff5207066cfef8eb380c35d398a884d357e2738d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/ff5207066cfef8eb380c35d398a884d357e2738d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
