)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~679804","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~I8492d52d8155e1bd75f99deb4fd19d445a831816","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"ansible-role-uwsgi","hashtags":[],"change_id":"I8492d52d8155e1bd75f99deb4fd19d445a831816","subject":"Start using uWSGI role","status":"MERGED","created":"2019-09-03 15:09:22.000000000","updated":"2019-09-06 15:49:28.000000000","submitted":"2019-09-06 15:47:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679804-1567784850531-b95e921d","meta_rev_id":"fd96b6697f217b6a5d2fa4f1ae2711df9d77f7f5","_number":679804,"virtual_id_number":679804,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-09-05 15:58:31.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 15:47:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-09-06 02:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-06 13:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 13:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 02:09:15.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-09-06 13:50:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-06 15:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cd6f977452b920ec92a08c2b27f4aedd4bb6fbf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-03 15:09:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8551cf76fb1045b4a68290bda55789c379cc64c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:18:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8dd37d36d9924426b18ebbc45a1b0ce3 : FAILURE in 1h 07m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4828a491136d4fa1acdf635189dd8d13 : FAILURE in 2h 03m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a6c6b679e7b34875973ca1dec2417734 : FAILURE in 53m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a5d668904534613b82633d933be1e9b : FAILURE in 1h 17m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2654b35bc8641cdbce82398ffc143a9 : FAILURE in 1h 14m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d78d03f829dd4cd0b62201a6bc264062 : FAILURE in 34m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc82a54a5f884fd8b6557958604feb15 : FAILURE in 59m 11s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d6a0f5ce065140588c21eb775dbd7821 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f11d1d3a49a4bc8ad46bc395a339ea5 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3b375de2eefbc20858d39e3ae0a46252dde50a5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 10:36:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8784dcd04cca47f28416bdb314e1c78225723b00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 11:26:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d38fbd74d9de400d84a7f8c7ccb3a673 : FAILURE in 30m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1b2e3555c1b64406a420c327926c7781 : FAILURE in 44m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c81e91b5655d41809d0774bb7d3e4586 : RETRY_LIMIT in 6m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d60e2142d6b44e99a8dd30d704dcbf87 : FAILURE in 27m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28fac1ee74834c63ad124bfd6b7913d9 : FAILURE in 33m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d43966306d1a45019b20c4b3f15be938 : RETRY_LIMIT in 4m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25a207a442c746c8a2c84dec9e1c8e13 : FAILURE in 22m 47s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4b2f37c7b6eb4c9fa5e6c0f04fdb9de2 : FAILURE in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f6fafb2f1f4b7b8d130fb97a93055b : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e5588c627c490bcee1ca91b5a1e5b950a37a44ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 11:51:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4c16b20ffd7811a27db9c43427f44a294650a49","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 12:33:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f1793bf236ce39a1ee8eaf2cb022d1d07f8bec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:25:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1284f4fa3b104bbbb0c0f0500d076c4e : FAILURE in 31m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f5a85ff89474153a3e245ea74d9f33d : FAILURE in 45m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5f06c6847d9b40809389bdf81b486afe : RETRY_LIMIT in 4m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5edd3dfba8547da9ee4e83e3e12587e : FAILURE in 30m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/69439bf395fb47af915f4ca0183b3b70 : FAILURE in 27m 06s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1082f1306138456b9f30e637e03825cf : RETRY_LIMIT in 4m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e97dc49a9c1446eea4892da10e19a3fb : FAILURE in 27m 51s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/823abcd818c849728e339f9af225e555 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f92bb9696824cafa2e9c48a01079537 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"6e8e3c79e5b0b96a7ce5f9b822dd721a4dee3a65","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 13:39:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b539cd6732007c4ccfb2854201a8ef6aed1d1143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 15:30:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ebe6d979494d4de59df859517de1542b : SUCCESS in 1h 17m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/06b4068cd2034e8db19c2cac5552d113 : SUCCESS in 1h 43m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cc9499ddee0e497b89f2e85ab2313b1a : RETRY_LIMIT in 4m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/643ab8bea0564320b060444ed8fe3a38 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/08b08537a2874462aaedb97da9694040 : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/74325741c8a040588a881066f78d4202 : RETRY_LIMIT in 7m 36s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e07e44bd045472393fa6b1453ac236e : SUCCESS in 57m 29s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c837a02069b0402288ba5cdaccf7d40a : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96380a9fbcb04643a8ccdbe3f78d1417 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":5},{"id":"bc7729e09e132ebe8e1795efbd69da409723ec97","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 15:58:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3662e86e177fd7a7c90220da0e9cae8f570320e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:08:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bef3f095b8c64555bdc434777126c5f4 : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/688378bbcd8c4e618a0284ea4ec144ea : SUCCESS in 1h 34m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/296baa1775c244d6a412c081dca57e2c : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c28356cfcc44c7b8a56ed60f9c9ddcd : SUCCESS in 1h 15m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7092c203491f462e9a07f3e79f567f17 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e05e732890ad412fa5e1564f9990cad7 : FAILURE in 29m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a26a04f886cd4267a6bf938c1691ef4c : SUCCESS in 47m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c2376e7029a24494b4e2492cdc10f81f : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e04f15235b4a0dabb4bc78ee36eec0 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":5},{"id":"8da6f89de05fb63b0c97c31616aecb19ea0648d9","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-06 02:09:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"154c48a9e4ba7555fbec40e7cf5ab9b88a04faba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-06 13:50:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e64a8a966bdcd82693b79824edd44cc414dfde15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:50:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e94298c01f1ebda2d9eb34cf70f7bc241ccaf0bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:47:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"10d4e8c73f2a611dcc3609d2ff6ab5a5b4cf4e92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:47:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6a89df872e554d858d78823084b8e8ae : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9f0a52ea2eb3458aa2f56d732119348a : SUCCESS in 1h 42m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f7eabe307f714166aafd8101e9152478 : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f930ecf937d24a0ea576eaea8b11e3fb : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d144b2e055840ffbba847809f236365 : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/374655ab5ea240d78deff2b257a8275d : SUCCESS in 48m 03s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/43735ad9736f4f879ae4d0748b2e6976 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/facac825947b4864a56cf6bb181ff1b9 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":5},{"id":"106e770cd8550fc3402e2c7e8e5a2a1a8d293360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:49:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6786fd8377d42a4b4835ee8d4549479 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"51c031147808adc69336d235e6b46a8d98860fdc","revisions":{"82666e4bb4645e6ccc76dcb05f3596c0bb182bcf":{"kind":"REWORK","_number":1,"created":"2019-09-03 15:09:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/679804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/04/679804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/1"}}},"commit":{"parents":[{"commit":"1f1f34942d0ea817d1ab1b9747502f3256247a29","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/1f1f34942d0ea817d1ab1b9747502f3256247a29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: I8492d52d8155e1bd75f99deb4fd19d445a831816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/82666e4bb4645e6ccc76dcb05f3596c0bb182bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/82666e4bb4645e6ccc76dcb05f3596c0bb182bcf"}]},"branch":"refs/heads/master"},"a40f9de6897b5d0d32164bb3956d3629b33605c7":{"kind":"REWORK","_number":2,"created":"2019-09-04 10:36:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/679804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/04/679804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/2"}}},"commit":{"parents":[{"commit":"1f1f34942d0ea817d1ab1b9747502f3256247a29","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/1f1f34942d0ea817d1ab1b9747502f3256247a29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 10:36:22.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: I8492d52d8155e1bd75f99deb4fd19d445a831816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/a40f9de6897b5d0d32164bb3956d3629b33605c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/a40f9de6897b5d0d32164bb3956d3629b33605c7"}]},"branch":"refs/heads/master"},"2b183bd537c9a17f91dd20792fe07383df6dc1d8":{"kind":"REWORK","_number":3,"created":"2019-09-04 11:51:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/679804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/04/679804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/3"}}},"commit":{"parents":[{"commit":"1f1f34942d0ea817d1ab1b9747502f3256247a29","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/1f1f34942d0ea817d1ab1b9747502f3256247a29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 11:51:06.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: I8492d52d8155e1bd75f99deb4fd19d445a831816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/2b183bd537c9a17f91dd20792fe07383df6dc1d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/2b183bd537c9a17f91dd20792fe07383df6dc1d8"}]},"branch":"refs/heads/master"},"b19b6963ba4441f18da67ecf1a9f0924d418ae9f":{"kind":"REWORK","_number":4,"created":"2019-09-04 12:33:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/679804/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/04/679804/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/4"}}},"commit":{"parents":[{"commit":"1f1f34942d0ea817d1ab1b9747502f3256247a29","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/1f1f34942d0ea817d1ab1b9747502f3256247a29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 12:33:01.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: I8492d52d8155e1bd75f99deb4fd19d445a831816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/b19b6963ba4441f18da67ecf1a9f0924d418ae9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/b19b6963ba4441f18da67ecf1a9f0924d418ae9f"}]},"branch":"refs/heads/master"},"51c031147808adc69336d235e6b46a8d98860fdc":{"kind":"REWORK","_number":5,"created":"2019-09-04 13:39:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/04/679804/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/04/679804/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/04/679804/5"}}},"commit":{"parents":[{"commit":"1f1f34942d0ea817d1ab1b9747502f3256247a29","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/1f1f34942d0ea817d1ab1b9747502f3256247a29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-03 15:09:07.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 13:39:05.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: I8492d52d8155e1bd75f99deb4fd19d445a831816\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/51c031147808adc69336d235e6b46a8d98860fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/51c031147808adc69336d235e6b46a8d98860fdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
