)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~712876","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~Iacea0be2be45a2c4b7d4f224de54fdece1516e4d","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"fix","hashtags":[],"change_id":"Iacea0be2be45a2c4b7d4f224de54fdece1516e4d","subject":"Missing document start \"---\"","status":"MERGED","created":"2020-03-13 09:17:36.000000000","updated":"2020-03-19 18:13:10.000000000","submitted":"2020-03-19 18:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712876-1584641243158-0b1117a9","meta_rev_id":"e8a1867d462169498ffc93c646f08e13c67e48e3","_number":712876,"virtual_id_number":712876,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 18:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 16:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-19 16:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 16:16:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-19 18:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"539031bdee5c988374a45047f17b6ceec89c03ac","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-03-13 09:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7caf75dfa075d66f3f1d53ecf7c6ba7f9c0254c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:30:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f896768ea7a64c8290fdeb8583642f63 : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0612747c8dd647f8a75da30bf25c4628 : SUCCESS in 2h 02m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/89766de72d924b0486b36903f7e20679 : SUCCESS in 1h 16m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5f0d012252143d69e7ee55e33317b47 : SUCCESS in 1h 04m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1919893d73e840bcb80a367d4bc95482 : FAILURE in 7m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b62d3f2c9ba4ee6a72525c3d7676330 : SUCCESS in 2h 09m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e42551e217744af681e6f1f4827c5ec3 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ea2e87b2abf64ac6b1ee850c832aa246 : FAILURE in 34m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f541f6ec50ad427db72be564fa8f6ee1 : SUCCESS in 51m 58s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17dfaa256a524df28e0fbfb10542431e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0539ca736d74752b47103ee93c86bc8 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"77a5650ff761ad9b853068e01e9276b84c4e13af","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-19 16:16:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c21fbbd827803dee266fc6d1d98cedb8d80413ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:17:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4139e099c751674d789b7c6fb96d5e2abdbab28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:07:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ebd2542d0f6042b79e710f766a882a62 : SUCCESS in 1h 05m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf9fcc75c1d547be86ca48869b34a183 : SUCCESS in 1h 31m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/729937dd1d0947c6838fac0ec07fbe8a : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8deb3443458462894c2b5938c101cea : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dfefd0de504e4e038d40497aeb6b3db8 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdab01c80b654cca86dd9ce1f7812a59 : SUCCESS in 57m 00s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ebef5460b5514c4f9c94fb856f475cca : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645f1ff29315414b967f8ff8cfbe55dc : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"42c98b4aef4ce1c7f8a76121e8222947fccd03cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:07:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0690586fa21512a73074723607ee213cfc1ed563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:13:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832df84c187b42eca45f10682b549f26 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7b1fb9da029a5b170205ea351cb8b25ae5548ed","revisions":{"d7b1fb9da029a5b170205ea351cb8b25ae5548ed":{"kind":"REWORK","_number":1,"created":"2020-03-13 09:17:36.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/76/712876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/76/712876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/76/712876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/76/712876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/76/712876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/76/712876/1"}}},"commit":{"parents":[{"commit":"7a5c18dfb896ae8d1776d75926ef340157b7e523","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/7a5c18dfb896ae8d1776d75926ef340157b7e523"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-03-13 09:16:58.000000000","tz":480},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-03-13 09:17:31.000000000","tz":480},"subject":"Missing document start \"---\"","message":"Missing document start \"---\"\n\nChange-Id: Iacea0be2be45a2c4b7d4f224de54fdece1516e4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d7b1fb9da029a5b170205ea351cb8b25ae5548ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d7b1fb9da029a5b170205ea351cb8b25ae5548ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
