)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~719601","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~I7ed46d3303795d036dd89268f1077e10d9ddd948","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I7ed46d3303795d036dd89268f1077e10d9ddd948","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-13 16:42:42.000000000","updated":"2020-04-14 09:11:25.000000000","submitted":"2020-04-14 09:07:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719601-1586855274314-61dee02a","meta_rev_id":"1107a043a6cfc6efc35989b1cc85646ddb2cedf7","_number":719601,"virtual_id_number":719601,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 09:07:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-04-13 17:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-14 08:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-14 08:50:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 17:04:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-14 08:50:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-14 09:07:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da845f7fee90d72c9e0eb9b0c31826201af2c35f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 16:42:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989694c207c050275fa2bad0e1ad451076382e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 16:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/44bb406d8c0a488dbdd6fec87b70a111 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9687c5c183dd4b6aaac187e6eb8c92fc : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6e36bac41b4980ad49da1420fef8d8 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"203b3ff75d487ab2ea399f771edde7a73960a21f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-13 17:04:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe1ee319d4b310a0d4347532d77388cf0b02911d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-14 08:50:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"803b36687eb594edc9c90d0e6521c5b357c000aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:51:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b14aa7838855d2733c40d6c4d830fc341295427","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:07:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bef044195d84bb735dddbc73b573f9ba9ac7ae3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:07:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e30ae60883bd4545ba77acab7355cb04 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a0e2b63c1640de874cd294f8ff4c48 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2cedf17d86d40e69671995ead8d98dc : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1bb768364d31a36c4310b69087ecde5945d50fbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb0a6370084434f8f3fc08dc62d75c2 : SUCCESS in 2m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc4442ff36ed4c11bf773fce164c45f8 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a691d380d3ca215c87f156054fbaa4bf7047575","revisions":{"5a691d380d3ca215c87f156054fbaa4bf7047575":{"kind":"REWORK","_number":1,"created":"2020-04-13 16:42:42.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/719601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/01/719601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/01/719601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/01/719601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/01/719601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/01/719601/1"}}},"commit":{"parents":[{"commit":"503b25617aa394d0410cc4169e3ade73e9ee9330","subject":"Merge \"Missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/503b25617aa394d0410cc4169e3ade73e9ee9330"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 16:39:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 16:42:13.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Remove obsolete sections from setup.cfg\n- Cleanup doc/source/conf.py to remove now obsolete content.\n- Enable warnings for docs build\n\nChange-Id: I7ed46d3303795d036dd89268f1077e10d9ddd948\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/5a691d380d3ca215c87f156054fbaa4bf7047575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/5a691d380d3ca215c87f156054fbaa4bf7047575"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
