)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~746574","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~stable%2Fussuri~Icb253467cd4f1304c7a9290ea8dc6677fd18af59","project":"openstack/openstack-ansible-os_barbican","branch":"stable/ussuri","hashtags":[],"change_id":"Icb253467cd4f1304c7a9290ea8dc6677fd18af59","subject":"Remove py2 packages","status":"MERGED","created":"2020-08-17 17:48:24.000000000","updated":"2020-08-18 10:35:17.000000000","submitted":"2020-08-18 10:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746574-1597746814746-c5357c5d","meta_rev_id":"c01365f21c72ee216f0e2a5ad9fdd90d86a9b462","_number":746574,"virtual_id_number":746574,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 10:33:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-08-18 08:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-08-18 06:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-08-18 08:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 06:18:03.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-18 08:51:50.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-08-18 10:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0917f6d53b1cd5ccc2891e30a9c31b59f5baef7e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-17 17:48:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c00c7ce10337e6c636afa4edf6ee1b194beff8ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 21:06:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6b37ca59068a48fbae58fc29f2cdfeaa : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e7828401ed3b492188b1585b8632d386 : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/325226e8138c4a44b18013353b55fdf2 : SUCCESS in 1h 43m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35d7e520a4c24bb9b932529aa2cca6e8 : TIMED_OUT in 3h 02m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd81cb41400f4ba9b7bb850feb1ebf91 : SUCCESS in 1h 18m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e579d2255b214a9f9b3538b77cd10032 : SUCCESS in 2h 35m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7bf34294b5d4ce28e84efd8c38ed743 : SUCCESS in 2h 09m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/572cb0403f1942eba330f4fbe33972e9 : SUCCESS in 1h 00m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92c6ec8ffdbf4eeda6a1dded611fb311 : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a6d08d3a0574d5c95bc63a5bb975d1e : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf8debdfeac04cc0bb2f70f837e1a2b0 : TIMED_OUT in 3h 03m 23s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/094d7f2ce2cb4e959df17ff75e97818f : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64f807b2e9a4726acc2b0df04bc3eb3 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ec352a52d98f38bc1b48c05b2d7a0bbb0023920d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-18 06:18:03.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73701e7952cbc77ebea70d9d84d0e215a824cdc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e0591e07f42a41a4b6312b8d17998873 : SUCCESS in 48m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/524514531b524a05bf0f3f9f01849b07 : SUCCESS in 1h 19m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e82fe6ac6294dcea15eb850b023f3f0 : SUCCESS in 1h 14m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ce53ea719174d0792e3515c2c3c825e : SUCCESS in 1h 21m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20f0da1d627466a8517517ce86a3f14 : SUCCESS in 1h 04m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/316fcbc8352140f993da73a9d8cc44c4 : SUCCESS in 2h 20m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f22623f62034bcca96cdbc5275e7449 : SUCCESS in 1h 52m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e4c1446b5f54305b2ef6b40bb42fa75 : SUCCESS in 59m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d5f406ec7067405495e661c164434bc1 : SUCCESS in 51m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7523f2d4f1f54e138c385595588857bb : SUCCESS in 41m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/406cef631ca345f8954ec2921e0c600e : SUCCESS in 50m 51s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/889a755d17124c2f8099d8487d22711b : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a0c214759a42ad9d37649fb594bb43 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"354fde586318e8fcdc9a866c10bb2c244ad6cdb5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-08-18 08:51:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7420935f947d1c48ef4f3e3796ff7b044efb0d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 08:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"595366069a6304ba82d8ccda3e9759da4b26f18d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:33:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8cc79e6567e8f2b3bcc51241d80e098843fff8e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:33:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ccec999c5fd24e4dbbf4573b63815775 : SUCCESS in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9648d636031242dbb94e4c6aa3afe5c2 : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0ece0528aabe4d39ae83f5c1552108f7 : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/434c5037f70c4e8896301dd5c383aed6 : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0ca4533ee0462ea70eedbe39ffdeda : SUCCESS in 1h 01m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5853b90aee9f4f1c9ff957395f0c7018 : SUCCESS in 55m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7d329aee0d244247b439088cb1ef112e : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ffce99f619246ac97d444042c3f69ff : SUCCESS in 52m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2570623f55434df783320b7a18e1d79e : SUCCESS in 50m 18s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/522cb5e1dbbb483bb15b4db0b0ce79a9 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a063d7839ac4a21a3fe3a5fd1120419 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c01365f21c72ee216f0e2a5ad9fdd90d86a9b462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 10:35:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9e71ea72b2401e954e291a059ccf7f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd66b509d458580a454fe5544a3af113049b53a9","revisions":{"dd66b509d458580a454fe5544a3af113049b53a9":{"kind":"REWORK","_number":1,"created":"2020-08-17 17:48:24.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/74/746574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/74/746574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/74/746574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/74/746574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/74/746574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/74/746574/1"}}},"commit":{"parents":[{"commit":"fb8f9ecded0f6ee779de01f2091ed18342c8f567","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/fb8f9ecded0f6ee779de01f2091ed18342c8f567"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-14 22:33:10.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-08-17 17:48:24.000000000","tz":0},"subject":"Remove py2 packages","message":"Remove py2 packages\n\nUbuntu Focal do not even have py2 packages, so we need to drop them,\nas anyway we use py3 everywhere.\n\nChange-Id: Icb253467cd4f1304c7a9290ea8dc6677fd18af59\n(cherry picked from commit 00d3bee5845f12d0233c8977b38ce3d29540925d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/dd66b509d458580a454fe5544a3af113049b53a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/dd66b509d458580a454fe5544a3af113049b53a9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
