)]}'
{"id":"openstack%2Fopenstack-ansible-os_barbican~895693","triplet_id":"openstack%2Fopenstack-ansible-os_barbican~master~Ia51971c077cef647c3d4e07d6cbc14b7bac70788","project":"openstack/openstack-ansible-os_barbican","branch":"master","topic":"osa/quorum_queues","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-10-03 16:53:21.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-03 16:53:21.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-10-02 09:33:30.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ia51971c077cef647c3d4e07d6cbc14b7bac70788","subject":"Add quorum support for service","status":"MERGED","created":"2023-09-18 14:44:24.000000000","updated":"2023-10-03 16:54:45.000000000","submitted":"2023-10-03 16:53:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895693-osa/quorum_queues","meta_rev_id":"3876d17a9a63548b6bcd49b0dc05f37c5858b0cd","_number":895693,"virtual_id_number":895693,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 16:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-09-28 14:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-02 09:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-03 14:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 17:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 17:20:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-09-21 16:59:11.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"b1950cf6fe4d1d059b1364f879fc91e8312f00d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-18 14:44:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"948a6e73670b80d6d49a78b044f1160bd3442562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd020332ca39445bad0cb0489df34c89\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/413f15b284274c1c8a4c0796ba3d33f6 : SUCCESS in 1h 17m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7d1e99b56d4aec8d420812404427f0 : SUCCESS in 1h 28m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cd70e22b63a34b07bc75159e2b62cc79 : SUCCESS in 1h 46m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd26575f67c641d582ea3e0108e16beb : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15f57cfc5f3e47eda68f41a4f805e411 : SUCCESS in 1h 41m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a38ddac0567a498094c3b78dd34b4e94 : SUCCESS in 1h 41m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c76a9057f834639a389f310b97e843d : SUCCESS in 2h 27m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98f4dadafc374f3db18ddfca3130072d : SUCCESS in 2h 11m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/800b666228444381a844a935c18f6c3d : SUCCESS in 1h 01m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09cfca79e7894c93ad894a51d8f3f82d : SUCCESS in 55m 45s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/943731e240bc481097024cd49e94259b : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e940f7ee30574fa9a9fb00b9837570b3 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5988ecb0fc2725309142f3e69aa3abec0d0fd334","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-20 17:20:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aae8d232b67fc430e772c30fd5b19999e57b8ad0","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-09-21 16:59:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51fc80c2c44fa356e98a60e915d7988f6cae73cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 17:01:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a710779d869158f55b597ad4bd9836321576f97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 19:02:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38b2c8d326b7410dba4803b0adfc9155\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6461ce3c4c8a4e1db47d704ccc8904da : SUCCESS in 1h 31m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8cfb908c655f45fc9e678752089b4e21 : SUCCESS in 1h 46m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd017760f59645bba16ea53156cc1b21 : SUCCESS in 1h 39m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53e1673785034b1faeee00fcf221bcde : FAILURE in 1h 52m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15ceef7db05149409f613911d4ff7dbe : SUCCESS in 1h 59m 40s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6d1959e56cc4a01b5e5163ac343bc76 : FAILURE in 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edaf16123483422ebe10ff978ae9b035 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5b27eabc8674e74212863e1c8aa75a9f1e11a48d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-26 15:30:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d75d365733f7651cff0332eae471ca74a1e0f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 17:56:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/425fa98124654216ae359dad523a7218\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4cb10d1c38a64b81ac6ee6f4017f71d7 : SUCCESS in 1h 38m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d80c0a4f1c2f412a93e2ff07b40825a5 : SUCCESS in 1h 38m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c246a6ce7533404984efdbb31b75cd12 : SUCCESS in 1h 51m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40df895ee4a44ee9a23d5e14c202af2f : SUCCESS in 1h 44m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f3b6e180c8e44b0c87a97af5214d4bca : SUCCESS in 2h 17m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed5ef39bd5264515ac2da96da231976c : SUCCESS in 1h 53m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b803950b870943549463cd20f02d5ba0 : SUCCESS in 2h 06m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74b37a0328a0489da90fd480ee0e3560 : SUCCESS in 2h 03m 16s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/573300b9e8224604a16e177a5cae8d7e : SUCCESS in 1h 07m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01bc945708fc4e83ae9a8820229187ab : SUCCESS in 1h 03m 03s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e74bd66190a74307a0fd532a0af6854f : FAILURE in 14m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f96c895a6ab4d6eb4f28c5a4b07eb72 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8305be892c923eb84521f053e3b3a603647dc28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-09-28 14:25:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e88565520ac4653a88597e125643a14b2ee7ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:22:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e0e4dbc915040febb27a470ef751194\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/65c11be6da4e4f6f9821a020c873a943 : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7691bb769024b6cb5744c9fba439872 : SUCCESS in 1h 36m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bfdc8131ff1a4d25a8b02e6c2707a370 : SUCCESS in 1h 52m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acd233aa910440f980767d22e9d4e25b : SUCCESS in 1h 59m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/97d1725f627d4b0f89215997f802451f : SUCCESS in 1h 59m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84a25ce9302b45e089409653743d7892 : SUCCESS in 1h 49m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c74aa6e6428749b6a4b48db728c5edd3 : FAILURE in 24m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c4faf8fb9eb45d394bad45a2d99e56f : SUCCESS in 2h 54m 47s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/254b956b01674948a62c7909907c6e79 : FAILURE in 16m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81bc7129b1024bfeb1b0fb705e29cc4c : SUCCESS in 2h 16m 41s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c37179e23dd54190a7ae06a3b5c8794f : SUCCESS in 1h 13m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd7e1030acbf467ea9e0ac3cad0bc9f8 : SUCCESS in 1h 05m 02s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5206f86f9725466398a0c8cc0a3c5c6c : SUCCESS in 15m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/323e3127c8b343fa9718ad1f053e805d : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1d2da9dde736c15c79f6e9f59eb9fcb857c87550","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-09-29 08:09:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab6ecbf207324d0d10772875c7a5229697e21e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:45:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfb8ff1c189f4f4ba4e6a359a7364184\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1b9cbb92f9a34d29883c0f49b9520f2c : SUCCESS in 1h 45m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f478e669d05442f886d6fc954cb88f70 : SUCCESS in 2h 01m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5550bf5782044d0693690cb6ee18a1ed : SUCCESS in 1h 39m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f015343e89542a48733d44c363df394 : SUCCESS in 1h 53m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f830f78687b7460397abb2a015bd39d4 : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13180bec904b469386e06026f2dea569 : SUCCESS in 1h 44m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/25914c0767fa4529af45084b38d884f2 : FAILURE in 2h 04m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c690aaa7047b4844a65b98d75467b30f : SUCCESS in 2h 21m 51s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad1f3c38354140d5a0b2c28ab677e32d : SUCCESS in 2h 24m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f379283d665e487d8b637d21a1e0beaa : FAILURE in 25m 00s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d76bfbe1812454bb766c29f1b6f31ce : FAILURE in 57m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32788ff974dd4f50ba1e9cd0e65c8922 : SUCCESS in 58m 47s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa614f8680314f8b830e69712862542f : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9e6b108935451390006f90a4863a0f : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"92e8601edb465f340e05e388588945c54c1703bb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-02 09:33:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1f79920be898993db9b14a180a96902c44bb4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 12:41:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ec381ea689440fba99262fcb452706\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d7362a8103044e34a2f5354e9f7a382b : SUCCESS in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/162bc8dbdc1d4a719165ac359eca97ca : SUCCESS in 1h 52m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/61ac2f3644054b95bf16651f8d5eef2f : SUCCESS in 1h 59m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2079658c10e4df98bf1cdb50c731146 : SUCCESS in 1h 43m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad756bcdb79d4cc69f496d5132d207b1 : SUCCESS in 2h 00m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74f9efa1aba84ac99d03187c1887a301 : SUCCESS in 1h 45m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df9bffbb3d4248c5aec7212475686d3e : SUCCESS in 3h 00m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36b5d91fba6d4ac299b03777d0647678 : SUCCESS in 2h 17m 58s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b3883d290a784b879c0681da2652e617 : SUCCESS in 2h 26m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d32778f0ff444fa7d4cc8b5796e621 : SUCCESS in 2h 06m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/00c862c90afd4b3797f57298ab10873f : SUCCESS in 1h 11m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c24011b75a9d48ef9d77d23a368eea55 : SUCCESS in 59m 01s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ed4c7b9d84644bdb902e9adcc057388 : SUCCESS in 13m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44577fc6e8f040bd92da239ddced3b05 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"963f7004cb1ef86d8448cfc0bce91e825aede235","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-03 14:29:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c74ddf5fcd6a20f7fcd4c01d7bf44531d08b51b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:32:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"226190639df2d650b7f327a965cc8906f8b928f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:53:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf416e660f34ba0ae4e900ee40a0880\n\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c17718605404331aa3d2fc15a90abac : SUCCESS in 1h 22m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b38241b0328c4d42b1cbd5b40eee3932 : SUCCESS in 2h 02m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbbee3de07cc4854a121b779958dbb32 : SUCCESS in 1h 45m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b1002a884f61490aafb183adb8491015 : SUCCESS in 2h 02m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59a91641f16f447c9d44d41ae684334f : SUCCESS in 1h 46m 34s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91dfb5782a794d7295e9fe2a87cde19e : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00572f4cd944d77bf6b5d261564c237 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"da935616b5ff06a38917e72f2f4d43f090fde8a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:53:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3876d17a9a63548b6bcd49b0dc05f37c5858b0cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 16:54:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e157298e95934012889fcaf058b87b19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4afac9a17a44922b2468d4911cc0d5e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea5e1adf63ee9f1cd3334d1e0b3288edc9cd7894","revisions":{"bb1d01377a6129dbaf9a5491af410119ed3c9aad":{"kind":"REWORK","_number":1,"created":"2023-09-18 14:44:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/895693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/93/895693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/1"}}},"commit":{"parents":[{"commit":"d461cff5dbe95a3cefb81b38ccc2dacef56f67ea","subject":"Merge \"Use proper galera port in configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/d461cff5dbe95a3cefb81b38ccc2dacef56f67ea"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-18 14:44:15.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-18 14:44:15.000000000","tz":120},"subject":"Add quorum support for service","message":"Add quorum support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nChange-Id: Ia51971c077cef647c3d4e07d6cbc14b7bac70788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/bb1d01377a6129dbaf9a5491af410119ed3c9aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/bb1d01377a6129dbaf9a5491af410119ed3c9aad"}]},"branch":"refs/heads/master"},"ea5e1adf63ee9f1cd3334d1e0b3288edc9cd7894":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-28 14:25:08.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/93/895693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_barbican","ref":"refs/changes/93/895693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_barbican refs/changes/93/895693/2"}}},"commit":{"parents":[{"commit":"3cd93dc6f2e2bc128355549dd083ea1478af4b2a","subject":"Fix example playbook linter errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/3cd93dc6f2e2bc128355549dd083ea1478af4b2a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-09-18 14:44:15.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-09-28 14:25:08.000000000","tz":0},"subject":"Add quorum support for service","message":"Add quorum support for service\n\nThis change implements and enables by default quorum support\nfor rabbitmq as well as providing default variables to globally tune\nit\u0027s behaviour.\n\nChange-Id: Ia51971c077cef647c3d4e07d6cbc14b7bac70788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/ea5e1adf63ee9f1cd3334d1e0b3288edc9cd7894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_barbican/commit/ea5e1adf63ee9f1cd3334d1e0b3288edc9cd7894"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
