)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~658402","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~master~I457c8e42ded7fd60523b79d8924f4c7dd153a995","project":"openstack/openstack-ansible-os_blazar","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I457c8e42ded7fd60523b79d8924f4c7dd153a995","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-10 15:46:38.000000000","updated":"2019-05-20 15:36:27.000000000","submitted":"2019-05-20 15:36:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658402-1558366587731-16b933f0","meta_rev_id":"9efe334bad614beec76fce434c9be5e28ef076c1","_number":658402,"virtual_id_number":658402,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-20 15:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-20 13:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-20 13:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-05-20 13:50:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 00:06:57.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-16 19:25:59.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-05-20 13:37:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-20 13:50:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-20 15:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ee8763c9948231841bf7e786617c749486f94f8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-10 15:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67dffd1c25614dd07fbe65e391d7df746f0d0bcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:20:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/1/check/openstack-ansible-linters/3bd09bf/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal-debian-stable/ffdb646/ : SUCCESS in 57m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal-centos-7/2810754/ : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/139f1bb/ : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/560a4ed/ : SUCCESS in 54m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/01c5019/ : FAILURE in 47m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/02/658402/1/check/openstack-tox-docs/546760d/html/ : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal_blazar-debian-stable/ea3f62b/ : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/02/658402/1/check/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/626dfb1/ : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bfb209cbf3caf186d01a2c87f96dc8c835210e67","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 10:12:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3fac7670c07f77e859b2a6c668f7db66f8de852c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:20:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/2/check/openstack-ansible-linters/03a88ae/ : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/2/check/openstack-ansible-deploy-aio_metal-debian-stable/61f1921/ : SUCCESS in 59m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/2/check/openstack-ansible-deploy-aio_metal-centos-7/51320ee/ : FAILURE in 1h 01m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/cf5c9ca/ : FAILURE in 55m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8f0d762/ : SUCCESS in 59m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/658402/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/7093502/ : FAILURE in 38m 31s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/02/658402/2/check/openstack-tox-docs/6819dfa/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ff5085d1929dfe3c3994c1da639d5307debf81f9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-15 12:26:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a09a4cc27a24538a359806bd9b4e8ee9b4cd89d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 13:58:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/3/check/openstack-ansible-linters/1cd9a6a/ : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-debian-stable/99c62df/ : SUCCESS in 58m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-centos-7/3a68272/ : SUCCESS in 1h 26m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/d46d261/ : SUCCESS in 1h 09m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/765d220/ : SUCCESS in 56m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/110762d/ : FAILURE in 37m 53s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/02/658402/3/check/openstack-tox-docs/16536aa/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":3},{"id":"1ded48e0d359b340c3e02debdf3ad809fbfcfaa0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-15 17:27:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"137ea1b87dfb7248744ad81124d0d35090ce28d3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-16 00:06:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7030d7f3ed248cc952ff509f266b40e792036741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 13:12:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7f273e0f70dbf5774ab829360498b2171b9eda89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:40:07.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/3/gate/openstack-ansible-linters/e058c34/ : SUCCESS in 4m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/fd606af/ : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/3/gate/openstack-ansible-deploy-aio_metal-centos-7/7a57d17/ : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/5a8f5e7/ : SUCCESS in 1h 09m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5cd29f6/ : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/658402/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5c319fe/ : FAILURE in 47m 56s\n- openstack-tox-docs http://logs.openstack.org/02/658402/3/gate/openstack-tox-docs/a52c2c8/html/ : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3},{"id":"5f54b7e70dd28028741801f26a9750fd71e7bb86","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-16 16:07:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cbfd3cc5d1623bdf97770a4584af435801d9ccf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 18:07:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/3/check/openstack-ansible-linters/8a70caf/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-debian-stable/926066a/ : SUCCESS in 1h 01m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-centos-7/3be5a92/ : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/74606e6/ : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b3da200/ : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/02/658402/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9f1a9c8/ : FAILURE in 56m 30s\n- openstack-tox-docs http://logs.openstack.org/02/658402/3/check/openstack-tox-docs/6ade374/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":3},{"id":"35e0201b11d438e9473ceee0e89bee19e612841c","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-16 19:25:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d431303a44a2aa8bac6c90630166690a2fe8feb0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 09:26:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b0b8085be9a94fa5f32c084ca304559852f5596f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 09:38:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"429d40dd42e87459cf6791e27e040bda264a538d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:08:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/4/check/openstack-ansible-linters/3d3f073/ : SUCCESS in 5m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/4/check/openstack-ansible-deploy-aio_metal-debian-stable/8939921/ : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/4/check/openstack-ansible-deploy-aio_metal-centos-7/0d4cbaa/ : SUCCESS in 1h 23m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/12021ed/ : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0b0e36f/ : SUCCESS in 51m 49s\n- openstack-tox-docs http://logs.openstack.org/02/658402/4/check/openstack-tox-docs/c3eb50c/html/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":4},{"id":"1c2be1ffb1b68931e7046b1c41d0666b5b62e841","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 13:37:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"da15267cf01bb570a3ff187e2e0fb98dae696553","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-20 13:50:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e725c916cf2054446ade67f534ed02be3e632e0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:50:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"984367ae0ae92e48bd4e5c524db1c6593ac7fe5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:36:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"529f45567b427924778a86643fb12d364e0b0061","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:36:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/02/658402/4/gate/openstack-ansible-linters/241396e/ : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/02/658402/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/037639a/ : SUCCESS in 1h 03m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/02/658402/4/gate/openstack-ansible-deploy-aio_metal-centos-7/5849e7d/ : SUCCESS in 1h 23m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/02/658402/4/gate/openstack-ansible-deploy-aio_metal-opensuse-150/ff209d9/ : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/02/658402/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/99d9197/ : SUCCESS in 1h 01m 43s\n- openstack-tox-docs http://logs.openstack.org/02/658402/4/gate/openstack-tox-docs/3df88cb/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7b55702bc7b79ec71246050a82e29e4ae0dba305","revisions":{"235946f639cb9f8adec7e9b79f7de33373dab42f":{"kind":"REWORK","_number":1,"created":"2019-05-10 15:46:38.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/02/658402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/02/658402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/1"}}},"commit":{"parents":[{"commit":"3439e7826ddc11d03843ecc1247f384739606b5f","subject":"Correct notification driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/3439e7826ddc11d03843ecc1247f384739606b5f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:29:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:46:22.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWhile this role isn\u0027t immediately exercised (the playbook\nis skipped), it is good enough to add coverage of the templates,\nshould we adapt the template in a global fashion.\n\nAdditionally, it is a problem to have different jobs\nfor SUSE support, as we want to use the integrated jobs template\nas a mean to get consistent across roles, so that\nit\u0027s easier to monitor jobs.\n\nChange-Id: I457c8e42ded7fd60523b79d8924f4c7dd153a995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/235946f639cb9f8adec7e9b79f7de33373dab42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/235946f639cb9f8adec7e9b79f7de33373dab42f"}]},"branch":"refs/heads/master"},"56b3910fadbf8f681866be7537b4ee881133a2b3":{"kind":"REWORK","_number":2,"created":"2019-05-14 10:12:32.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/02/658402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/02/658402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/2"}}},"commit":{"parents":[{"commit":"3439e7826ddc11d03843ecc1247f384739606b5f","subject":"Correct notification driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/3439e7826ddc11d03843ecc1247f384739606b5f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:29:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 10:11:46.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso tests the blazar role in the metal jobs template.\n\nChange-Id: I457c8e42ded7fd60523b79d8924f4c7dd153a995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/56b3910fadbf8f681866be7537b4ee881133a2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/56b3910fadbf8f681866be7537b4ee881133a2b3"}]},"branch":"refs/heads/master"},"732e12b4424b9d7c7f86258db3c4f78534d1ba7d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-15 12:26:50.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/02/658402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/02/658402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/3"}}},"commit":{"parents":[{"commit":"0d44c8e0732f68eaec1648b0eda3db986a13fa50","subject":"Add SUSE and CentOS support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/0d44c8e0732f68eaec1648b0eda3db986a13fa50"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:29:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-15 12:26:50.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso tests the blazar role in the metal jobs template.\n\nChange-Id: I457c8e42ded7fd60523b79d8924f4c7dd153a995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/732e12b4424b9d7c7f86258db3c4f78534d1ba7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/732e12b4424b9d7c7f86258db3c4f78534d1ba7d"}]},"branch":"refs/heads/master"},"7b55702bc7b79ec71246050a82e29e4ae0dba305":{"kind":"REWORK","_number":4,"created":"2019-05-20 09:38:38.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/02/658402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/02/658402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/02/658402/4"}}},"commit":{"parents":[{"commit":"0d44c8e0732f68eaec1648b0eda3db986a13fa50","subject":"Add SUSE and CentOS support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/0d44c8e0732f68eaec1648b0eda3db986a13fa50"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-10 15:29:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 09:38:06.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso tests the blazar role in the metal jobs template.\n\nChange-Id: I457c8e42ded7fd60523b79d8924f4c7dd153a995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7b55702bc7b79ec71246050a82e29e4ae0dba305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7b55702bc7b79ec71246050a82e29e4ae0dba305"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
