)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~658588","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~stable%2Fstein~Id2879da97d576d135c3966e7fb7ca48aa1332ea2","project":"openstack/openstack-ansible-os_blazar","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id2879da97d576d135c3966e7fb7ca48aa1332ea2","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:58:19.000000000","updated":"2019-05-12 02:59:23.000000000","submitted":"2019-05-12 02:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658588-1557629963869-1a85bf1c","meta_rev_id":"54d103278ef31852ccbc8796a0af022d421020cc","_number":658588,"virtual_id_number":658588,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 02:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:56:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:56:23.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 02:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7296efccd1bab3b9f2c5e2303f21ad9f326c9c58","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0c0516f5e501b281e6dea97c8c42990df6a2071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/658588/1/check/openstack-ansible-linters/56f64a4/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/88/658588/1/check/openstack-tox-docs/e364bdf/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/88/658588/1/check/build-openstack-releasenotes/7708537/html/ : SUCCESS in 3m 11s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/88/658588/1/check/openstack-ansible-deploy-aio_metal_blazar-debian-stable/0c1fa1f/ : SUCCESS in 1h 02m 52s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/88/658588/1/check/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/7e1dc68/ : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c4c4a6f9a6475f7a809e6cfd67349a7bb9f703f9","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:56:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4dc506b46fa2e66752a36fcf13a03c4ff2553aeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:56:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3056422ff035542ff03e014c2d2b187405a43b0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:59:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52c88c23d32f445f1b3a640f3e2a774c02945163","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:59:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/88/658588/1/gate/openstack-ansible-linters/cc37dc3/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/88/658588/1/gate/openstack-tox-docs/f2c9df8/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/88/658588/1/gate/build-openstack-releasenotes/b95ea05/html/ : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/88/658588/1/gate/openstack-ansible-deploy-aio_metal_blazar-debian-stable/77e28ca/ : SUCCESS in 56m 54s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/88/658588/1/gate/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/67dceb1/ : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6632ec12062a46ba06f49a9ad8218bd41d19d533","revisions":{"6632ec12062a46ba06f49a9ad8218bd41d19d533":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:58:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/658588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/88/658588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/88/658588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/88/658588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/88/658588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/88/658588/1"}}},"commit":{"parents":[{"commit":"2e68a9d90a8266fc95beb0617f47b44fd39a517b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/2e68a9d90a8266fc95beb0617f47b44fd39a517b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:18.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id2879da97d576d135c3966e7fb7ca48aa1332ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/6632ec12062a46ba06f49a9ad8218bd41d19d533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/6632ec12062a46ba06f49a9ad8218bd41d19d533"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
