)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~666697","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~stable%2Fstein~I2ae110ed0a8e9119dbe2e9d310bdf8387a3b32f8","project":"openstack/openstack-ansible-os_blazar","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2ae110ed0a8e9119dbe2e9d310bdf8387a3b32f8","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:03:48.000000000","updated":"2019-07-15 21:40:07.000000000","submitted":"2019-07-15 21:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666697-1563226807365-cc061412","meta_rev_id":"82c7c7062ff54d71d810d5982414dfe1f017e772","_number":666697,"virtual_id_number":666697,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-15 21:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 21:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-24 09:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 09:10:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-15 21:40:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5deaaad6503cf16251ae3a7ec816d8bd87e03457","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:03:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e284ed0cda9ad196febd47b3aa60e50d2143b20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:30:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/666697/1/check/openstack-ansible-linters/6029081/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/97/666697/1/check/openstack-tox-docs/192ae05/html/ : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/97/666697/1/check/openstack-ansible-deploy-aio_metal_blazar-debian-stable/a594014/ : SUCCESS in 1h 03m 44s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/97/666697/1/check/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/dc18fd1/ : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2ac737baa2aa85a8d6dd33244f15b8ad176a18fb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 21:37:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b061e00eb69681ff2d3d71b4e729e1b2fdc82988","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-24 09:10:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"756a8c079e3bc84b88615b4fc1d5fc18003d928f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 09:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd20d613ce6090f6adc12fea0281f6690f9c5133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:19:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/97/666697/1/gate/openstack-ansible-linters/e57991e/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/97/666697/1/gate/openstack-tox-docs/c260155/html/ : SUCCESS in 3m 12s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/97/666697/1/gate/openstack-ansible-deploy-aio_metal_blazar-debian-stable/aa9c96b/ : SUCCESS in 1h 08m 08s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/97/666697/1/gate/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/a5caca4/ : FAILURE in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a7928cebf687e351b2b989a6237ba850f2f9e737","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 18:59:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea5a0cda087814d755615113c5b5be4578447509","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:19:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/666697/1/check/openstack-ansible-linters/608d695/ : SUCCESS in 8m 16s\n- openstack-tox-docs http://logs.openstack.org/97/666697/1/check/openstack-tox-docs/88e6f4f/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/97/666697/1/check/openstack-ansible-deploy-aio_metal_blazar-debian-stable/916e5ff/ : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/97/666697/1/check/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/500b18d/ : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"28820855d1706c6f20c95b7f15e69a774ce11d37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:20:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fcacb1b14c879d6f70f76dcf50ea47017a9888c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:40:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/97/666697/1/gate/openstack-ansible-linters/af33cd7/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/97/666697/1/gate/openstack-tox-docs/5a10870/html/ : SUCCESS in 4m 00s\n- openstack-ansible-deploy-aio_metal_blazar-debian-stable http://logs.openstack.org/97/666697/1/gate/openstack-ansible-deploy-aio_metal_blazar-debian-stable/b20b6fa/ : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic http://logs.openstack.org/97/666697/1/gate/openstack-ansible-deploy-aio_metal_blazar-ubuntu-bionic/23a7fae/ : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"02e569615a2aacb5cb2e0d57d002fb9c1f6e04f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:40:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c8ebe204e66048f257dd4469ef04e621990d7b2","revisions":{"5c8ebe204e66048f257dd4469ef04e621990d7b2":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:03:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/666697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/97/666697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/97/666697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/97/666697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/97/666697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/97/666697/1"}}},"commit":{"parents":[{"commit":"6632ec12062a46ba06f49a9ad8218bd41d19d533","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/6632ec12062a46ba06f49a9ad8218bd41d19d533"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:03:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:03:46.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2ae110ed0a8e9119dbe2e9d310bdf8387a3b32f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/5c8ebe204e66048f257dd4469ef04e621990d7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/5c8ebe204e66048f257dd4469ef04e621990d7b2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
