)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~709481","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~master~Ibcbfd3dfc1dc043afd2d5ef7160db09341497e41","project":"openstack/openstack-ansible-os_blazar","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ibcbfd3dfc1dc043afd2d5ef7160db09341497e41","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:08:31.000000000","updated":"2020-02-26 08:42:20.000000000","submitted":"2020-02-26 08:40:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709481-1582706442845-8d25ec41","meta_rev_id":"dcb054b8aa44a52782cfc6e28585e2e46ed29011","_number":709481,"virtual_id_number":709481,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 08:40:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 07:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-26 07:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 07:01:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-26 08:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb8b8a807885fd89dc72d815465e8d2db0479321","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:08:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42b2370e5e713d5ad0ac791cffe55df3578b6326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:02:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac73e20f54da489a8c6e68e0ebdc5530 : SUCCESS in 8m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d0b27a84b7e84bf083903db722d4408a : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2675469b186548e6bcef3d4bee42ce6f : SUCCESS in 1h 42m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/602fa40d19ad44bc8c93f8c319f5aa36 : SUCCESS in 1h 43m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52cc5aff744242fb929470fcb8bae81e : SUCCESS in 1h 29m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aab1652d837646d0a40c7ba6d145ac34 : FAILURE in 1h 21m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adee9544e00b4876801f23755a80fb3f : FAILURE in 1h 07m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1372639d019f41838b1c1e07773ee127 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed546e287c34c3082bb884dadce7473 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2f7efeb248373e5c29c3d67294a077574194baf8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:51:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68581b5d413c885d40d07c189c0b16204b60419a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:00:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5a9868e3d0b4ce2b4d07045b9fb3528 : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cdf8f3303ca44f92b0fe22bbda71fd75 : FAILURE in 1h 06m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/90caf7acb5174864b54d05303c224037 : SUCCESS in 1h 47m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40306f916b6e4323b6bf63c34a2bcf27 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/969b8555a41e49709366d455f261e477 : SUCCESS in 1h 25m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f10c85eee4d74fe88ebc2b0c2dccc849 : FAILURE in 1h 27m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bedcf0765644ef29af05f6eca9725ea : FAILURE in 1h 06m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c14e561a64479e890bb84af0285e90 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522a96f1202f4713b3b0c4158ae736c8 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73e6939895ee1ca6e4ed1a1cd0214110b6c5fd91","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 21:18:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c6551cddd3450568a8762c5cb52e51cac7dac9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 23:08:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3b03a054ae5b4fd6a2039c513a35bfc7 : SUCCESS in 8m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/06eac980076542f78b62d78125e0f192 : SUCCESS in 1h 12m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f48bc23511614bd2919aa531e16a094e : SUCCESS in 1h 47m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f4a832725ec048538d7f9afbd2a70d5c : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26ec29399f4f4aa38b19620830d31bce : SUCCESS in 1h 15m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9124a21679d7494dbeac468354b22f20 : FAILURE in 1h 29m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f6a160cbe4c4b1d87616684d8448432 : FAILURE in 1h 02m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be093fbcb55e49ad852e89a24833efbe : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de37e269725e4a81bc3fdf297af34dda : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bd28687c6d6366a2929dd52350fefe46b2a22611","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-26 07:01:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbeaa5c265997a3b22be0b15b90e976e5e469b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 07:01:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b1363dfdfbf320dcd5d9b29346c76761f6d0917","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:40:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4acfb8e0204ac23f0d8ce9ed279c018aa222d0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:40:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/75bbc7ee4c6544a9b4a1d711d4080586 : SUCCESS in 6m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c8b4ca19cd264234a369c60a87406be1 : SUCCESS in 1h 09m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4024248b3c814e4bb20d0d7f08733b78 : SUCCESS in 1h 37m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a3cf81d3a42d4ce3ae87deda3a974cdb : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4368a46eea3467393365925d3d814cc : SUCCESS in 1h 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc589b521ddf427da75fa5befaab3644 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cb2f1b6701a47cfa216e9cb401becd3 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4b276d6ecf6c326a8c50a170a35968d75c3c7a77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:42:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5c6b288b7347258a4b362840984583 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38da7d9aeecc4fb695b2c522a76fc58a : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0e37be5c1fd76f65a0a19eed9501f22712c7e18","revisions":{"e0e37be5c1fd76f65a0a19eed9501f22712c7e18":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:08:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/709481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/81/709481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/81/709481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/81/709481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/81/709481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/81/709481/1"}}},"commit":{"parents":[{"commit":"155fa917a38fe55364d9bc06620fd2880fc0a335","subject":"Merge \"tox: Keeping going with docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/155fa917a38fe55364d9bc06620fd2880fc0a335"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:08:27.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ibcbfd3dfc1dc043afd2d5ef7160db09341497e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/e0e37be5c1fd76f65a0a19eed9501f22712c7e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/e0e37be5c1fd76f65a0a19eed9501f22712c7e18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
