)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~732244","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~stable%2Fussuri~I2e58958cf3efadb85ec8e848f33b6be9c077469e","project":"openstack/openstack-ansible-os_blazar","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I2e58958cf3efadb85ec8e848f33b6be9c077469e","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-06-01 13:12:16.000000000","updated":"2020-06-01 18:29:08.000000000","submitted":"2020-06-01 18:27:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732244-1591036053071-69ea3769","meta_rev_id":"6a7357480e0d6206cb64eb466753c28c3089eb6d","_number":732244,"virtual_id_number":732244,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-06-01 18:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-06-01 18:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 18:17:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-01 18:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"746c13052b73ed6f7406cf2e0f6c425e1b9499e1","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-01 13:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03734412f94671faa93795f98732a05e5fe01b24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:56:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f6dfdcd9c11442a1a1928fbde48f11b1 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c2b9b691114f159dc69ac1f113d54b : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/605d39b964774e16baf92d73921c3332 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b06111943102d63064808e26dc31156895ac033e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-01 18:17:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8db2bacad410a36c8ee7ace5874609de707c5947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:17:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f15cc515260932cd664509fc802db7e38025ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:27:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/64bbdc7859f645b39c479d4f07f1311b : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e6b5c005ac4c9d9697874a803981aa : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5551273416564b5db3d66431c194750f : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4fd8b257bf4b0af2045be7eb966a83103bc667bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:27:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d2fd4813971bc833e63b294678698259335ff517","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:29:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ec5858d966438bb3d4a9965ba0d3df : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b10876ab59a4daea263d0698fe461c1 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f248f29188050093617baf9acc08b4a967eca6f7","revisions":{"f248f29188050093617baf9acc08b4a967eca6f7":{"kind":"REWORK","_number":1,"created":"2020-06-01 13:12:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/44/732244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/44/732244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/44/732244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/44/732244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/44/732244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/44/732244/1"}}},"commit":{"parents":[{"commit":"4265a99afe0ca33754896cb25c97415c16e384ac","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/4265a99afe0ca33754896cb25c97415c16e384ac"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:12:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:12:14.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I2e58958cf3efadb85ec8e848f33b6be9c077469e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/f248f29188050093617baf9acc08b4a967eca6f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/f248f29188050093617baf9acc08b4a967eca6f7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
