)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~880652","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~master~Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99","project":"openstack/openstack-ansible-os_blazar","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-05-03 11:17:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99","subject":"Add TLS support to blazar backends","status":"MERGED","created":"2023-04-17 15:05:15.000000000","updated":"2023-05-03 11:18:36.000000000","submitted":"2023-05-03 11:17:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880652-tls-backend","meta_rev_id":"dd04ed7a44eaba76814fd3bc08236bd73b982a99","_number":880652,"virtual_id_number":880652,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 11:17:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-02 10:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 09:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-03 09:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 17:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 10:47:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-05-03 09:29:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"39056e500c965f70fb321c7f0295289b5005267a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-17 15:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5ee2ddcb31861c01206893890aea737593406d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-17 15:09:45.000000000","message":"Topic set to tls-backend","accounts_in_message":[],"_revision_number":1},{"id":"8d75425e4891fdddb3b9ce3adc089388809011fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:19:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cfadeb8fbfc4e67ab911d13cc9cfb47\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a3093b5b3c49bfb134adf723e58b50 : SUCCESS in 11m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/aa05e82d722c4d84adb09ac112e03b28 : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1261f20850b423e974f671d97668ba2 : SUCCESS in 1h 32m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c711b769f9e4e46965f467ace956682 : SUCCESS in 1h 22m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41b29e1a398c4820bf73463b577e1327 : SUCCESS in 1h 39m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bff9b12b50634f80b449f996eabdc833 : SUCCESS in 1h 56m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fdede08ab8194219bbb24556d2292685 : FAILURE in 1h 57m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23cd1cd551c14ead88e0017ea57aba0d : FAILURE in 2h 07m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b527dd918f4247d888a4d16c87b1e4d9 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"38cecee6be10c618a8be397a6946ab8924ed808b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-18 11:06:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6fa17e8d6f22c0e6a338cb2dfe20202be5b053bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 13:43:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585dfb1440f745cea59a17553411231d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e721b51559645d0bb908e53e3214c8b : SUCCESS in 11m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/35696cb4d5b14fc1a3871fd4aa65df61 : SUCCESS in 1h 38m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a38078fd8f446408af1d78547aaea16 : SUCCESS in 1h 31m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19b0e34b1dae4d43a6bf7b100bfd170e : SUCCESS in 1h 41m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20b5b26c35045e7843f9ab9b267c0e3 : SUCCESS in 1h 43m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ca4d5291b9347dca530aed2546013f9 : SUCCESS in 1h 31m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf4e29159b8a4f94a2260e4b66cc1c86 : SUCCESS in 2h 28m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/099cbf383aa140889783cad47c74538f : SUCCESS in 2h 17m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7919acce6d4319a9d820bbf09a73c5 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5581acac3c73746fb0adccb8b7ea52495ee76201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 10:48:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d948d0fc5a11eab9736ea5bf5562cc9c00392f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 13:12:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33040f96f8dc403697740dd8ec52dc45\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85d79d7073af47648bef31a8c6e47a9d : SUCCESS in 10m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f93bfc7067fb4ec3930d8f046474e5df : SUCCESS in 1h 29m 03s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0a89036bc104b30b06cf00166f97c3a : SUCCESS in 1h 46m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6276858fca8141da803e6fb829222432 : SUCCESS in 1h 55m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/910871045400492890f2bfa2e53dfcc7 : SUCCESS in 1h 52m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e09bb6186e946f2b15d4c3e33c7dcdf : SUCCESS in 1h 37m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a367733eaeb043b9b3b925f55e58d1f6 : SUCCESS in 2h 15m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/523bf0eafe0b4df5b6dec83584ed7ead : SUCCESS in 2h 17m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0acc544458e748c68924419b580cf3ac : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5bf35870f1a53b830f8cd2d3a28b977dd938ceac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 22:00:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9f8669ad51f7365627d4745f47b796276bba611a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 00:29:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d42d9542714ce1be5d5d1a82f19561\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34f44eb24c404158bbf3d5b5a80f0cae : SUCCESS in 10m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/930979077acb4df894db4517fcd216c6 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ae3b599137e43e599c6dadc7b0c87d8 : SUCCESS in 1h 55m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/53cb16148a544a258a246a22136aa191 : SUCCESS in 1h 32m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19de13b937cb45189a885e23acf63467 : SUCCESS in 1h 40m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b07629e8e1ef48798ff2ae0af8c49000 : SUCCESS in 1h 27m 58s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/527db987bc9f48199b6c23d489a4a678 : SUCCESS in 2h 16m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33062d7869b54b4bb43d468d7d543628 : SUCCESS in 2h 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ac3aad7f07447cbdda8d8350683b50 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a4ffe6ce78ca67c483fb610185d56fcdc9cd28ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-29 16:35:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"58d3bddf543964c42adba9ab9fa2e9a1897c3062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 18:34:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7d7cee902ac4839a5fe59116ba074df\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920c2e7ac7c04a85bc92f8f293d0b1b2 : SUCCESS in 11m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fce60255e7214ac5ad29d8b102160245 : SUCCESS in 1h 34m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5022a78e9a14b4d89810b678ef4ddb2 : SUCCESS in 1h 36m 56s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/372653da7b9841c9bb45f708a2c486fd : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebdc3962040842b8a817715a25f5c378 : SUCCESS in 1h 40m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5776aeb52304d4cadb1c5f6c2aacb93 : SUCCESS in 1h 22m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85982c662f8143f1b15292d66f67ed93 : FAILURE in 1h 38m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d7d9b4d07df4b38a28f38b0d3bbb1c4 : SUCCESS in 1h 55m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f246a5d9495a448d8d2c8c28691727de : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0b3b96951d739200496e124ba5e4ba8d105a4384","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-29 23:31:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5d09d6608d27fcadb16e4f4eafd913cf26b31b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-30 02:00:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0145c555f704c2cb5a8484785416687\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4f8ec2e32d4d1095013ae4f2194d6e : SUCCESS in 10m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9c4dc896a38e45489b82ffa9c6106a13 : SUCCESS in 1h 33m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d1c873c5fc74b9f9f708e4beebd82d1 : SUCCESS in 1h 26m 49s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/298e296b273d475ca98587f69e380d86 : SUCCESS in 1h 36m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6ae4493faa54a76a8c0bcc97e08609c : SUCCESS in 1h 41m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a563147592c499ea8f89d2fd3b37030 : SUCCESS in 1h 28m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5381ec00a0c945cabc34c99a4e41e24c : SUCCESS in 2h 21m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/880ef220df7e4fb6be51acc6d744931a : SUCCESS in 2h 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bfba2fa5de46daba3cfaa690448a6d : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":5},{"id":"54746ffe05ffab975cf0a5aa9d908807631b6ddb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 10:47:34.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fd5d7ce20c98a2bb5d84b2248a368db649839afe","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-03 09:29:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"26f70b6ae83e08322bd0ee393c5c1d6f44d01efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:31:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"90a29dca1d6ced1b4507c0d5a6736d87b7f880d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:17:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a13c183d389e40e5aa0abf59d1ccb198\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/091399f733cc4953b7be07723c4921a5 : SUCCESS in 11m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a6b2f25322f04ad282be753bc53391b4 : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/139e06f36152480797f89e21e1d384c1 : SUCCESS in 1h 26m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/343d28ae9ef748bebe753e05ca292171 : SUCCESS in 1h 45m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc3972e0d464ee68dd986171e08a318 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"51dbd55fe296b53653f164c7e407f5897dd56e2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:17:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"dd04ed7a44eaba76814fd3bc08236bd73b982a99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:18:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bd052c036964b91a34ac15c70587405\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56656878d0554e378888c28222be5dbb : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8b39e84ec10f5ae4a3f6475ec2e7cef63796c566","revisions":{"3462741d842dfbb6b24f68a6ef666af1cf944604":{"kind":"REWORK","_number":1,"created":"2023-04-17 15:05:15.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/52/880652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/52/880652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/1"}}},"commit":{"parents":[{"commit":"2f583cdb7299c37b151cbc26800e014273321372","subject":"Add uWSGI support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/2f583cdb7299c37b151cbc26800e014273321372"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:01:41.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:04:52.000000000","tz":120},"subject":"Add TLS support to blazar backends","message":"Add TLS support to blazar backends\n\nBy overriding the variable `blazar_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the blazar backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/3462741d842dfbb6b24f68a6ef666af1cf944604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/3462741d842dfbb6b24f68a6ef666af1cf944604"}]},"branch":"refs/heads/master"},"7f491eda3810374faddaa889ca55f58055894b94":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-18 11:06:21.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/52/880652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/52/880652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/2"}}},"commit":{"parents":[{"commit":"042cd12cdcbecd227038b327e17d57970d985e5f","subject":"Add uWSGI support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/042cd12cdcbecd227038b327e17d57970d985e5f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:01:41.000000000","tz":120},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2023-04-18 11:06:21.000000000","tz":0},"subject":"Add TLS support to blazar backends","message":"Add TLS support to blazar backends\n\nBy overriding the variable `blazar_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the blazar backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7f491eda3810374faddaa889ca55f58055894b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7f491eda3810374faddaa889ca55f58055894b94"}]},"branch":"refs/heads/master","description":"Rebase"},"7f0d87764e6f7271733e1687623d48b3a37ecdac":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-24 10:48:54.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/52/880652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/52/880652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/3"}}},"commit":{"parents":[{"commit":"c9d10f41fad8bcee0f3604727007b8034a7fa67f","subject":"Add uWSGI support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/c9d10f41fad8bcee0f3604727007b8034a7fa67f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:01:41.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-24 10:48:33.000000000","tz":120},"subject":"Add TLS support to blazar backends","message":"Add TLS support to blazar backends\n\nBy overriding the variable `blazar_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the blazar backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7f0d87764e6f7271733e1687623d48b3a37ecdac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/7f0d87764e6f7271733e1687623d48b3a37ecdac"}]},"branch":"refs/heads/master"},"ae7f002389b443b4b73b6ce6b4afe21d5e129614":{"kind":"REWORK","_number":4,"created":"2023-04-24 22:00:19.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/52/880652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/52/880652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/4"}}},"commit":{"parents":[{"commit":"c9d10f41fad8bcee0f3604727007b8034a7fa67f","subject":"Add uWSGI support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/c9d10f41fad8bcee0f3604727007b8034a7fa67f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:01:41.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-24 22:00:12.000000000","tz":120},"subject":"Add TLS support to blazar backends","message":"Add TLS support to blazar backends\n\nBy overriding the variable `blazar_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the blazar backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/ae7f002389b443b4b73b6ce6b4afe21d5e129614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/ae7f002389b443b4b73b6ce6b4afe21d5e129614"}]},"branch":"refs/heads/master"},"8b39e84ec10f5ae4a3f6475ec2e7cef63796c566":{"kind":"REWORK","_number":5,"created":"2023-04-29 16:35:58.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/52/880652/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/52/880652/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/52/880652/5"}}},"commit":{"parents":[{"commit":"c9d10f41fad8bcee0f3604727007b8034a7fa67f","subject":"Add uWSGI support to blazar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/c9d10f41fad8bcee0f3604727007b8034a7fa67f"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-17 15:01:41.000000000","tz":120},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-29 16:35:56.000000000","tz":120},"subject":"Add TLS support to blazar backends","message":"Add TLS support to blazar backends\n\nBy overriding the variable `blazar_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the blazar backend api.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: Ic784cb180ff4cbc81c230b0a3a62015a71ea3f99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/8b39e84ec10f5ae4a3f6475ec2e7cef63796c566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/8b39e84ec10f5ae4a3f6475ec2e7cef63796c566"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
