)]}'
{"id":"openstack%2Fopenstack-ansible-os_blazar~950504","triplet_id":"openstack%2Fopenstack-ansible-os_blazar~master~If13a79c15a42fa52150fd3fee0207b6828e5cde3","project":"openstack/openstack-ansible-os_blazar","branch":"master","topic":"osa-wsgi-module","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-05-22 17:39:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If13a79c15a42fa52150fd3fee0207b6828e5cde3","subject":"Switch from wsgi script to wsgi module","status":"MERGED","created":"2025-05-21 09:33:11.000000000","updated":"2025-05-22 17:40:02.000000000","submitted":"2025-05-22 17:39:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950504-osa-wsgi-module","meta_rev_id":"5a618ca81df7a181db6471cf3b29ccb70a003e58","_number":950504,"virtual_id_number":950504,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-22 17:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-05-22 11:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2025-05-22 15:38:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2025-05-22 15:38:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 11:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-22 11:55:01.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-05-22 15:38:01.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"79a9945e1a4956553921f49706848e8f7683bec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-21 09:33:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed3678ac412cc7d8d32559fd9f3bdf4257d207f3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-21 10:47:06.000000000","message":"Topic set to osa-wsgi-module","accounts_in_message":[],"_revision_number":1},{"id":"2bd851bb7114f371c752a24515b0e055eed8667c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 11:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc6bbfad2054f86bde9c096bb674cb4\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d842b6f5632b4003a068baf7406cac22 : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/09cf9b65b5a44db7a201f646d794fa66 : SUCCESS in 1h 34m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0467f55f1b3442ee9761fa64781505c1 : SUCCESS in 1h 27m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/98e30787e2094b128837d8cfa9c455fa : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70950752a875427d9480a41600f29b71 : SUCCESS in 1h 28m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a15e7c4dca9e4c81a7d45333f56bb3cf : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c09374296c7f46fd8fa816f37e59ce40 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e592f5a8b2d4c5d81c39a9c71972ead : SUCCESS in 1h 32m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab281b05fad542129188248b663ede1b : SUCCESS in 1h 25m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/428746cafb2e4406a4014e298ed685bc : SUCCESS in 2h 04m 39s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/72a4ecbc127e4e3587d04b65a532ac01 : SUCCESS in 2h 07m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e8ad2f06dd46f482a736d054c96c61 : SUCCESS in 1h 45m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94b9a7cdb6e1429a9ec9d73a0a7310c8 : SUCCESS in 2h 19m 08s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dfba097fe0d64673befaf1761a0503cb : SUCCESS in 2h 12m 56s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76a4b35228714fdb845c6d2ec3ebbbde : SUCCESS in 1h 58m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a62268c2a4c4badad5dfe2c1a76c0c6 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c55fd1423a2c4cd6d851e5b4caa0f953ac9631ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-22 11:55:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2b40277d8d0698106c210424ca7680d120b5e2c","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-05-22 15:38:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d273457245519bbdeb8418dfcdfb64f218665f70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 15:40:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6ed02313b4a59d6f36ebf29d76fdfe2ff3d7558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:39:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73973efebcc24e65ad4bf09d68c2c916\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd3ab0c3a15946369748d75c7ae4aace : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f5cb6ae1e1584695900fbf8f376248b8 : SUCCESS in 1h 53m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/edb66acfd4384d7bba0e57f2e61969e3 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cb73f714a814b278a6c11e41c4ad69d : SUCCESS in 1h 23m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0beb1f5515045699480beea60b6f434 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c9887650bfc4172a47c2bc6020a0c79 : SUCCESS in 1h 33m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d837f095cbe43109c258e2127bb3323 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/770850ab489e4ad08c7889c9de484030 : SUCCESS in 1h 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad81d426e8d47f4b2c324516aaeb556 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d2060065949df1f645a77e3a404c943cb340c3d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:39:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a618ca81df7a181db6471cf3b29ccb70a003e58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bc601e05c7a49a2a9ee92f364d59330\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc225ac5ac2d4833a971df8900fd59c0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"863893a5227d460e7e0970844e2fcb6ac8438e80","revisions":{"863893a5227d460e7e0970844e2fcb6ac8438e80":{"kind":"REWORK","_number":1,"created":"2025-05-21 09:33:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/04/950504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_blazar","ref":"refs/changes/04/950504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/04/950504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/04/950504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/04/950504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_blazar refs/changes/04/950504/1"}}},"commit":{"parents":[{"commit":"13c0161b5de4c5ac9b0e1a69cf6b10cb58ee41ec","subject":"Auto-fix usage of modules via FQCN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/13c0161b5de4c5ac9b0e1a69cf6b10cb58ee41ec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 09:32:47.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-21 09:32:47.000000000","tz":60},"subject":"Switch from wsgi script to wsgi module","message":"Switch from wsgi script to wsgi module\n\nChange-Id: If13a79c15a42fa52150fd3fee0207b6828e5cde3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/863893a5227d460e7e0970844e2fcb6ac8438e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_blazar/commit/863893a5227d460e7e0970844e2fcb6ac8438e80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
