)]}'
{"id":"openstack%2Fopenstack-ansible-os_ceilometer~541275","triplet_id":"openstack%2Fopenstack-ansible-os_ceilometer~master~I20104cc357f78b01e2b1246a3f51d609bb2f4f24","project":"openstack/openstack-ansible-os_ceilometer","branch":"master","topic":"bug/1721554","hashtags":[],"change_id":"I20104cc357f78b01e2b1246a3f51d609bb2f4f24","subject":"Drop MongoDB installation and API file configurations","status":"MERGED","created":"2018-02-06 13:46:32.000000000","updated":"2018-02-07 16:23:49.000000000","submitted":"2018-02-07 16:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"541275-1518020629130-d987e295","meta_rev_id":"74313c593fa80d3efd405666fda3aa762ae3d77b","_number":541275,"virtual_id_number":541275,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2018-02-07 16:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2018-02-07 09:47:42.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2018-02-07 15:42:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2018-02-07 15:42:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-07 09:47:42.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2018-02-07 15:42:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2018-02-07 16:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"305a0edaea08ed7881392a57396ccc2dc9a69427","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-02-06 13:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61e82db496b5f83231913fc92b0cde3fd4ac96d8","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-02-06 13:49:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a2784ee57967bc0907eec276b4fbb9fb25422c1b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2018-02-06 20:29:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0c20ca11f3dc92411da08496c297b867ff70902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 01:02:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/75/541275/3/check/build-openstack-sphinx-docs/4a8377f/html/ : SUCCESS in 2m 20s\n- build-openstack-releasenotes http://logs.openstack.org/75/541275/3/check/build-openstack-releasenotes/d62980f/html/ : SUCCESS in 1m 40s\n- openstack-ansible-linters http://logs.openstack.org/75/541275/3/check/openstack-ansible-linters/b47c298/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-centos-7-nv http://logs.openstack.org/75/541275/3/check/openstack-ansible-functional-centos-7-nv/a236577/ : FAILURE in 27m 56s (non-voting)\n- openstack-ansible-functional-opensuse-423-nv http://logs.openstack.org/75/541275/3/check/openstack-ansible-functional-opensuse-423-nv/1cdd6d6/ : SUCCESS in 30m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/75/541275/3/check/openstack-ansible-functional-ubuntu-xenial/aa3648f/ : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1c21f52c9e44ae8ddea1fd9c5085b5ecf0b7f0e0","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-02-07 09:47:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"157f3e33ecaa7bf6dc2cb91d74265dde475ba078","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-02-07 15:42:02.000000000","message":"Patch Set 3: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1de6744ea88ac66c505febab9929977a62ce79b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 15:48:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35316027b9cafc94bf7d7e82048b3d74d97aee9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 16:23:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/75/541275/3/gate/build-openstack-sphinx-docs/56502ce/html/ : SUCCESS in 2m 45s\n- build-openstack-releasenotes http://logs.openstack.org/75/541275/3/gate/build-openstack-releasenotes/e08a82c/html/ : SUCCESS in 2m 23s\n- openstack-ansible-linters http://logs.openstack.org/75/541275/3/gate/openstack-ansible-linters/0fef6b5/ : SUCCESS in 6m 20s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/75/541275/3/gate/openstack-ansible-functional-ubuntu-xenial/0b47955/ : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":3},{"id":"636ec767a1d4835dcc2aee9fd405cdd5eee4e1ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 16:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33f5dfa7e3a5aaab9e26e99bd298371ad37d6958","revisions":{"66af883ee8db40422f3194a8c4a9d2b719fe5fb5":{"kind":"REWORK","_number":1,"created":"2018-02-06 13:46:32.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/75/541275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/75/541275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/1"}}},"commit":{"parents":[{"commit":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","subject":"Remove setting of version/release from releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 13:45:24.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 13:46:30.000000000","tz":-300},"subject":"Drop MongoDB installation and api-paste file configuration","message":"Drop MongoDB installation and api-paste file configuration\n\nCeilometer no longer stores any data so the MongoDB installation\nplaybook is not used.  In addition, it no longer has an API so\nthe api-paste file is useless as well.\n\nChange-Id: I20104cc357f78b01e2b1246a3f51d609bb2f4f24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/66af883ee8db40422f3194a8c4a9d2b719fe5fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/66af883ee8db40422f3194a8c4a9d2b719fe5fb5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","is_merged_in_target_branch":true,"change_id":"Ia73e43f262b36eb19b7854830c329e36e3807859","change_number":520937,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c05de65d44a101dbf79633b2731cae08f9e96161":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-02-06 13:49:21.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/75/541275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/75/541275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/2"}}},"commit":{"parents":[{"commit":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","subject":"Remove setting of version/release from releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 13:45:24.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 13:49:04.000000000","tz":-300},"subject":"Drop MongoDB installation and api-paste file configuration","message":"Drop MongoDB installation and api-paste file configuration\n\nCeilometer no longer stores any data so the MongoDB installation\nplaybook is not used.  In addition, it no longer has an API so\nthe api-paste file is useless as well.\n\nCloses-Bug: #1747612\nChange-Id: I20104cc357f78b01e2b1246a3f51d609bb2f4f24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/c05de65d44a101dbf79633b2731cae08f9e96161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/c05de65d44a101dbf79633b2731cae08f9e96161"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","is_merged_in_target_branch":true,"change_id":"Ia73e43f262b36eb19b7854830c329e36e3807859","change_number":520937,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"33f5dfa7e3a5aaab9e26e99bd298371ad37d6958":{"kind":"REWORK","_number":3,"created":"2018-02-06 20:29:02.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/75/541275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/75/541275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/75/541275/3"}}},"commit":{"parents":[{"commit":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","subject":"Remove setting of version/release from releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 13:45:24.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2018-02-06 20:28:48.000000000","tz":-300},"subject":"Drop MongoDB installation and API file configurations","message":"Drop MongoDB installation and API file configurations\n\nCeilometer no longer stores any data so the MongoDB installation\nplaybook is not used.  In addition, it no longer has an API so\nthe api-paste file is useless as well.\n\nCloses-Bug: #1747612\nChange-Id: I20104cc357f78b01e2b1246a3f51d609bb2f4f24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/33f5dfa7e3a5aaab9e26e99bd298371ad37d6958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/33f5dfa7e3a5aaab9e26e99bd298371ad37d6958"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"55bb04eaad4dd5c7fdad742b3557dc30dc9d45bf","is_merged_in_target_branch":true,"change_id":"Ia73e43f262b36eb19b7854830c329e36e3807859","change_number":520937,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
