)]}'
{"id":"openstack%2Fopenstack-ansible-os_ceilometer~557042","triplet_id":"openstack%2Fopenstack-ansible-os_ceilometer~master~I1e722ad353564a6586e5b597d73b5e1dad6033d9","project":"openstack/openstack-ansible-os_ceilometer","branch":"master","topic":"bp/python-build-install-simplification","hashtags":[],"change_id":"I1e722ad353564a6586e5b597d73b5e1dad6033d9","subject":"Use a common python build/install role","status":"MERGED","created":"2018-03-27 20:38:28.000000000","updated":"2018-09-03 12:30:22.000000000","submitted":"2018-09-03 12:30:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"557042-1535977822435-5c4522c2","meta_rev_id":"2a4063de630e0c2b355e6185790bd1fc338ef80a","_number":557042,"virtual_id_number":557042,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":2,"date":"2018-09-03 12:30:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2018-09-03 11:49:21.000000000","_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-09-03 10:56:25.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2018-09-03 11:49:21.000000000","_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-03 10:56:25.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2018-09-03 11:49:21.000000000","updated_by":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"reviewer":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"state":"REVIEWER"},{"updated":"2018-09-03 12:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c33f43822deede45475434e6024324e8aa50df1","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-03-27 20:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04cea8285d8c9f42ab7c16ac9af81d48a2653084","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-27 21:36:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/42/557042/1/check/build-openstack-sphinx-docs/ace81c0/html/ : SUCCESS in 2m 19s\n- openstack-ansible-linters http://logs.openstack.org/42/557042/1/check/openstack-ansible-linters/9251032/ : SUCCESS in 5m 25s\n- openstack-ansible-functional-centos-7-nv http://logs.openstack.org/42/557042/1/check/openstack-ansible-functional-centos-7-nv/4aecd5a/ : SUCCESS in 51m 44s (non-voting)\n- openstack-ansible-functional-opensuse-423-nv http://logs.openstack.org/42/557042/1/check/openstack-ansible-functional-opensuse-423-nv/4ad1fd0/ : SUCCESS in 36m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/42/557042/1/check/openstack-ansible-functional-ubuntu-xenial/8f54e2c/ : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f922f17e73ed853f760e1d439f538d94f6493298","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-03-28 07:37:13.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4ae30efb2039c4a984245fa2ea9a2c84c9cd70c5","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-04-20 14:04:09.000000000","message":"Abandoned\n\nAbandoning this for now - will come back to it another time.","accounts_in_message":[],"_revision_number":1},{"id":"1cd840a83ba4bca6751c85fd059e265da6e6ce3a","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-09-01 17:10:41.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a9458cd748ded5b4ce33b6f5944480c4878e9793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 17:10:53.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fdfe66b29130a884abcde74b8896ea8a357fa15a","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-09-01 17:15:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2ea106db790ac92874b93dfc3800e074754bed91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 17:55:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/42/557042/2/check/build-openstack-sphinx-docs/508fb2c/html/ : SUCCESS in 2m 13s\n- openstack-ansible-linters http://logs.openstack.org/42/557042/2/check/openstack-ansible-linters/52ec74e/ : SUCCESS in 4m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-centos-7/289f39b/ : SUCCESS in 38m 57s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-opensuse-423/de1500c/ : SUCCESS in 33m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-ubuntu-bionic/cad0c0c/ : SUCCESS in 30m 19s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-ubuntu-xenial/7c59725/ : SUCCESS in 27m 35s\n- openstack-tox-docs http://logs.openstack.org/42/557042/2/check/openstack-tox-docs/2ad10a9/html/ : SUCCESS in 2m 09s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-distro_install-ubuntu-xenial/2141eb1/ : SUCCESS in 25m 27s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-distro_install-centos-7/92d2e4d/ : FAILURE in 32m 28s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/42/557042/2/check/openstack-ansible-functional-distro_install-opensuse-423/b4fc9af/ : SUCCESS in 33m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b9470d2d010ab919be9ff40fa3cfd59d51a93f9a","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-09-03 09:47:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2fb08bb331db60451f729976317712c31d3627cc","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-09-03 10:17:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f84a01b98996c961e2678017c523d4d27c5b339f","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-09-03 10:24:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dccd7a90d5a7a4e7da5d3bae9c7271a11ba87726","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-09-03 10:40:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4d44e05e2099a6b8c75171f6bdbfafd941bd8b2f","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-09-03 10:56:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9792fbc4f9e3b601d5aceb5f69aabf987cffb9df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 11:24:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/42/557042/3/check/build-openstack-sphinx-docs/141fa8b/html/ : SUCCESS in 3m 09s\n- openstack-ansible-linters http://logs.openstack.org/42/557042/3/check/openstack-ansible-linters/83892cc/ : SUCCESS in 6m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-centos-7/f0825e5/ : SUCCESS in 43m 05s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-opensuse-423/012f2d8/ : SUCCESS in 32m 47s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-ubuntu-bionic/31ac39f/ : SUCCESS in 31m 19s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-ubuntu-xenial/226d73c/ : SUCCESS in 30m 51s\n- openstack-tox-docs http://logs.openstack.org/42/557042/3/check/openstack-tox-docs/5771865/html/ : SUCCESS in 2m 42s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-distro_install-ubuntu-xenial/3bbf577/ : SUCCESS in 31m 07s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-distro_install-centos-7/4112179/ : FAILURE in 30m 35s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/42/557042/3/check/openstack-ansible-functional-distro_install-opensuse-423/416a84f/ : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":3},{"id":"4faec66ea21e4e124b960021b6cabe2a968e8721","author":{"_account_id":23163,"name":"Markos Chandras (hwoarang)","email":"markos@chandras.me","username":"mchandras"},"date":"2018-09-03 11:49:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e2b72466f921d94028e69c1affc5b94e7d1e2b84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 11:49:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"33d96b2d41b773a474e966357a8a0b78e305e92e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 12:30:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"191009f16020223b7a83858d0c857da55c85c7a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 12:30:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/42/557042/3/gate/build-openstack-sphinx-docs/8d4f5a1/html/ : SUCCESS in 2m 47s\n- openstack-ansible-linters http://logs.openstack.org/42/557042/3/gate/openstack-ansible-linters/e97b707/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-centos-7/abed2b0/ : SUCCESS in 37m 32s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-opensuse-423/f214cfa/ : SUCCESS in 34m 46s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-ubuntu-bionic/1ce28db/ : SUCCESS in 33m 43s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-ubuntu-xenial/bce963e/ : SUCCESS in 31m 35s\n- openstack-tox-docs http://logs.openstack.org/42/557042/3/gate/openstack-tox-docs/2ba7cbd/html/ : SUCCESS in 2m 48s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-distro_install-ubuntu-xenial/ba2917f/ : SUCCESS in 30m 53s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/42/557042/3/gate/openstack-ansible-functional-distro_install-opensuse-423/f72bb3f/ : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e0502655ad9dcdf45c47d9c35196865e10d8d50","revisions":{"486f285eb012e61edadab95ede6dfd326cc26649":{"kind":"REWORK","_number":1,"created":"2018-03-27 20:38:28.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/42/557042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/42/557042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/1"}}},"commit":{"parents":[{"commit":"a4e93e419d7765d95c70240b102a290a3f1fd167","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/a4e93e419d7765d95c70240b102a290a3f1fd167"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-03-27 20:38:22.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-03-27 20:38:22.000000000","tz":60},"subject":"Use a common python build/install role for source-base deployment","message":"Use a common python build/install role for source-base deployment\n\nInstead of copying a common set of code between all the roles,\nswitch to using a common role which checks whether a deploy host\nalready has the appropriate venv package. If it does not, build\nit on the fly and pull it to the deploy host.\n\nImplementing this does away with the requirement to do builds\non the repo container. Once this has been implemented into all\nroles then the repo_build role will be retired.\n\nChange-Id: I1e722ad353564a6586e5b597d73b5e1dad6033d9\nImplements: blueprint python-build-install-simplification\nSigned-off-by: Jesse Pretorius \u003cjesse.pretorius@rackspace.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/486f285eb012e61edadab95ede6dfd326cc26649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/486f285eb012e61edadab95ede6dfd326cc26649"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a4e93e419d7765d95c70240b102a290a3f1fd167","is_merged_in_target_branch":true,"change_id":"Ibd01abb008b09014a892e8198253a503520bca81","change_number":556911,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3629e60c6c234e2e5727bb3d46ab2ea8665c4311":{"kind":"REWORK","_number":2,"created":"2018-09-01 17:15:26.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/42/557042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/42/557042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/2"}}},"commit":{"parents":[{"commit":"974f71785c5372a3ae5959a1b9db7bf6df8e1d02","subject":"switch documentation job to new PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/974f71785c5372a3ae5959a1b9db7bf6df8e1d02"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-03-27 20:38:22.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-09-01 17:14:54.000000000","tz":60},"subject":"Use a common python build/install role","message":"Use a common python build/install role\n\nIn order to radically simplify how we prepare the service\nvenvs, we use a common role to do the wheel builds and the\nvenv preparation. This makes the process far simpler to\nunderstand, because the role does its own building and\ninstalling. It also reduces the code maintenance burden,\nbecause instead of duplicating the build processes in the\nrepo_build role and the service role - we only have it all\ndone in a single place.\n\nWe also change the role venv tag var to use the integrated\nbuild\u0027s common venv tag so that we can remove the role\u0027s\nvenv tag in group_vars in the integrated build. This reduces\nmemory consumption and also reduces the duplication.\n\nThis is by no means the final stop in the simplification\nprocess, but it is a step forward. The will be work to follow\nwhich:\n\n1. Replaces \u0027developer mode\u0027 with an equivalent mechanism\n   that uses the common role and is simpler to understand.\n   We will also simplify the provisioning of pip install\n   arguments when doing this.\n2. Simplifies the installation of optional pip packages.\n   Right now it\u0027s more complicated than it needs to be due\n   to us needing to keep the py_pkgs plugin working in the\n   integrated build.\n3. Deduplicates the distro package installs. Right now the\n   role installs the distro packages twice - just before\n   building the venv, and during the python_venv_build role\n   execution.\n\nDepends-On: https://review.openstack.org/598957\nChange-Id: I1e722ad353564a6586e5b597d73b5e1dad6033d9\nImplements: blueprint python-build-install-simplification\nSigned-off-by: Jesse Pretorius \u003cjesse.pretorius@rackspace.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/3629e60c6c234e2e5727bb3d46ab2ea8665c4311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/3629e60c6c234e2e5727bb3d46ab2ea8665c4311"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"974f71785c5372a3ae5959a1b9db7bf6df8e1d02","is_merged_in_target_branch":true,"change_id":"I82b4afe5608ef1564aea242d82a89b0a39460383","change_number":594240,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6e0502655ad9dcdf45c47d9c35196865e10d8d50":{"kind":"REWORK","_number":3,"created":"2018-09-03 10:40:43.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/42/557042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ceilometer","ref":"refs/changes/42/557042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ceilometer refs/changes/42/557042/3"}}},"commit":{"parents":[{"commit":"974f71785c5372a3ae5959a1b9db7bf6df8e1d02","subject":"switch documentation job to new PTI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/974f71785c5372a3ae5959a1b9db7bf6df8e1d02"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-03-27 20:38:22.000000000","tz":60},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse.pretorius@rackspace.co.uk","date":"2018-09-03 10:40:38.000000000","tz":0},"subject":"Use a common python build/install role","message":"Use a common python build/install role\n\nIn order to radically simplify how we prepare the service\nvenvs, we use a common role to do the wheel builds and the\nvenv preparation. This makes the process far simpler to\nunderstand, because the role does its own building and\ninstalling. It also reduces the code maintenance burden,\nbecause instead of duplicating the build processes in the\nrepo_build role and the service role - we only have it all\ndone in a single place.\n\nWe also change the role venv tag var to use the integrated\nbuild\u0027s common venv tag so that we can remove the role\u0027s\nvenv tag in group_vars in the integrated build. This reduces\nmemory consumption and also reduces the duplication.\n\nThis is by no means the final stop in the simplification\nprocess, but it is a step forward. The will be work to follow\nwhich:\n\n1. Replaces \u0027developer mode\u0027 with an equivalent mechanism\n   that uses the common role and is simpler to understand.\n   We will also simplify the provisioning of pip install\n   arguments when doing this.\n2. Simplifies the installation of optional pip packages.\n   Right now it\u0027s more complicated than it needs to be due\n   to us needing to keep the py_pkgs plugin working in the\n   integrated build.\n3. Deduplicates the distro package installs. Right now the\n   role installs the distro packages twice - just before\n   building the venv, and during the python_venv_build role\n   execution.\n\nDepends-On: https://review.openstack.org/598957\nChange-Id: I1e722ad353564a6586e5b597d73b5e1dad6033d9\nImplements: blueprint python-build-install-simplification\nSigned-off-by: Jesse Pretorius \u003cjesse.pretorius@rackspace.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/6e0502655ad9dcdf45c47d9c35196865e10d8d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ceilometer/commit/6e0502655ad9dcdf45c47d9c35196865e10d8d50"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"974f71785c5372a3ae5959a1b9db7bf6df8e1d02","is_merged_in_target_branch":true,"change_id":"I82b4afe5608ef1564aea242d82a89b0a39460383","change_number":594240,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
