)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~614099","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I0beb18e2e0afc39923d5a1c12c4096af09becb66","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"I0beb18e2e0afc39923d5a1c12c4096af09becb66","subject":"missing document start \"---\"","status":"MERGED","created":"2018-10-30 07:36:13.000000000","updated":"2019-04-24 20:06:17.000000000","submitted":"2019-04-24 20:06:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614099-1556136377793-b16e9fe7","meta_rev_id":"e79cde5324b8beeadf4f17db2b888e9691b8e82a","_number":614099,"virtual_id_number":614099,"owner":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 20:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:29:20.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 20:06:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27ab24377f2ef713ec34b63551fbefd2fba7679a","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-30 07:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ac7fbd6bbd351ce16578267fd4319fa9af698a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-30 08:33:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/614099/1/check/openstack-ansible-linters/b5f23e0/ : SUCCESS in 6m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-centos-7/effb071/ : SUCCESS in 52m 32s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-opensuse-150/4636235/ : SUCCESS in 50m 31s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-opensuse-423/db7b89f/ : SUCCESS in 43m 17s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-ubuntu-bionic/21f6ff7/ : SUCCESS in 39m 39s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-centos-7/879487b/ : SUCCESS in 39m 53s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-opensuse-150/164b46f/ : SUCCESS in 43m 00s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-opensuse-423/819f679/ : SUCCESS in 28m 01s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/1f00207/ : FAILURE in 35m 11s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/99/614099/1/check/openstack-tox-docs/022e1eb/html/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"81c268ebcf70d874941ac9b47882de1a0b52eb59","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-11-08 03:55:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"fdb799b2be485bb4c396935aacd493aefdd92967","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-12-10 23:04:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9ed0d645b182f224f712d9ee2391ff72387d1417","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 00:36:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/614099/1/check/openstack-ansible-linters/5538fd2/ : SUCCESS in 5m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-centos-7/79c5383/ : SUCCESS in 43m 50s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-opensuse-150/4be40a7/ : SUCCESS in 47m 39s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-opensuse-423/efcb7e2/ : SUCCESS in 45m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-ubuntu-bionic/f776feb/ : SUCCESS in 40m 20s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-centos-7/39d6c70/ : SUCCESS in 32m 54s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-opensuse-150/6337762/ : FAILURE in 26m 16s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-opensuse-423/f6ad1b0/ : FAILURE in 20m 03s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/99/614099/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/77def7f/ : SUCCESS in 38m 51s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/99/614099/1/check/openstack-tox-docs/9eee6f2/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a29afcd3420bd399a8560ee7e8262c62b722f8aa","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:29:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84ecd8ec02092e8b67db6637783ce8437b666295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:29:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be393fb1bdcfcd534bafb638bf309107afcb8316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:06:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e79cde5324b8beeadf4f17db2b888e9691b8e82a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:06:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/614099/1/gate/openstack-ansible-linters/3d287c5/ : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/614099/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/cc267b1/ : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/614099/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6c46c9d/ : SUCCESS in 1h 32m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/99/614099/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/f670bab/ : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/614099/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7ef8911/ : SUCCESS in 1h 19m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/614099/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/06ef678/ : SUCCESS in 1h 02m 16s\n- openstack-tox-docs http://logs.openstack.org/99/614099/1/gate/openstack-tox-docs/74eeef5/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41c7eb2f2ad3baa7b3252d39a1d876f4cb3bdcff","revisions":{"41c7eb2f2ad3baa7b3252d39a1d876f4cb3bdcff":{"kind":"REWORK","_number":1,"created":"2018-10-30 07:36:13.000000000","uploader":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"ref":"refs/changes/99/614099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/99/614099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/99/614099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/99/614099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/99/614099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/99/614099/1"}}},"commit":{"parents":[{"commit":"981611b35a06253da43f506e2cc49129e47b2c8e","subject":"Set container_name as host_var in test inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/981611b35a06253da43f506e2cc49129e47b2c8e"}]}],"author":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:35:56.000000000","tz":0},"committer":{"name":"wangqi","email":"wang.qi@99cloud.net","date":"2018-10-30 07:36:11.000000000","tz":0},"subject":"missing document start \"---\"","message":"missing document start \"---\"\n\nChange-Id: I0beb18e2e0afc39923d5a1c12c4096af09becb66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/41c7eb2f2ad3baa7b3252d39a1d876f4cb3bdcff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/41c7eb2f2ad3baa7b3252d39a1d876f4cb3bdcff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
