)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~655519","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I57fccb5debbf3a53c9978d55915bda5691be796f","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"bug/1826250","hashtags":[],"change_id":"I57fccb5debbf3a53c9978d55915bda5691be796f","subject":"Adding cinder_service_timeout","status":"ABANDONED","created":"2019-04-24 18:25:08.000000000","updated":"2019-07-30 19:46:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e479c7214b8082eb7bead9d7ce202165fce4aee","_number":655519,"virtual_id_number":655519,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-04-25 00:52:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":-1,"date":"2019-04-26 17:35:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 00:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 17:35:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"2cb2f1e04c1b98fe576c16424b116a88998e864a","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 18:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc594af361a5195bfc557e870473085b25854f82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 20:02:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/655519/1/check/openstack-ansible-linters/40f5601/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/655519/1/check/openstack-ansible-deploy-aio_metal-debian-stable/620202f/ : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/655519/1/check/openstack-ansible-deploy-aio_metal-centos-7/e615cb2/ : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/655519/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/e218aad/ : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/655519/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/64a566b/ : SUCCESS in 1h 18m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/655519/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ea1b498/ : SUCCESS in 56m 44s\n- openstack-tox-docs http://logs.openstack.org/19/655519/1/check/openstack-tox-docs/6e7df29/html/ : SUCCESS in 2m 43s\n- build-openstack-releasenotes http://logs.openstack.org/19/655519/1/check/build-openstack-releasenotes/e8a03cc/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bd08c132bba839390e6e93321528d7c7b264a454","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 20:09:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d53cacdce9b92a9ddf428efe8cc733fb0357c967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 00:52:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/655519/2/check/openstack-ansible-linters/32c935e/ : SUCCESS in 5m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/655519/2/check/openstack-ansible-deploy-aio_metal-debian-stable/e1b4a45/ : SUCCESS in 1h 00m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/655519/2/check/openstack-ansible-deploy-aio_metal-centos-7/e9a5c84/ : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/655519/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/c654625/ : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/655519/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c7b67a2/ : SUCCESS in 57m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/655519/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/69669a9/ : SUCCESS in 1h 03m 42s\n- openstack-tox-docs http://logs.openstack.org/19/655519/2/check/openstack-tox-docs/0d85cfc/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/19/655519/2/check/build-openstack-releasenotes/f028d44/html/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"938c64e7f53cbeb1bb245de1b86cca88b72d258f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-26 17:35:03.000000000","message":"Patch Set 2: Code-Review-1\n\nplease review feedback on https://review.opendev.org/#/c/655559/1","accounts_in_message":[],"_revision_number":2},{"id":"8448f54e6e9d9e86e0d8f1512d9c41611eb69813","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:46:01.000000000","message":"Abandoned\n\nno follow up patch and adds an unnecessary variable which can just be used by an override","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd6f5d1926c2536100a9f8394344d0571ac2fc6d","revisions":{"20583f5bdeffcf8d2ed89dfd53115806f7dcb74b":{"kind":"REWORK","_number":1,"created":"2019-04-24 18:25:08.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/19/655519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/19/655519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/1"}}},"commit":{"parents":[{"commit":"e055d332acadc79cb18a3a202966596aea812163","subject":"Merge \"Revert \"Convert role to use a common systemd service role\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e055d332acadc79cb18a3a202966596aea812163"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 18:24:33.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 18:24:33.000000000","tz":-300},"subject":"Adding cinder_service_timeout","message":"Adding cinder_service_timeout\n\nThe override `cinder_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I57fccb5debbf3a53c9978d55915bda5691be796f\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/20583f5bdeffcf8d2ed89dfd53115806f7dcb74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/20583f5bdeffcf8d2ed89dfd53115806f7dcb74b"}]},"branch":"refs/heads/master"},"fd6f5d1926c2536100a9f8394344d0571ac2fc6d":{"kind":"REWORK","_number":2,"created":"2019-04-24 20:09:46.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/19/655519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/19/655519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/19/655519/2"}}},"commit":{"parents":[{"commit":"e055d332acadc79cb18a3a202966596aea812163","subject":"Merge \"Revert \"Convert role to use a common systemd service role\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/e055d332acadc79cb18a3a202966596aea812163"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 18:24:33.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:09:34.000000000","tz":-300},"subject":"Adding cinder_service_timeout","message":"Adding cinder_service_timeout\n\nThe override `cinder_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nChange-Id: I57fccb5debbf3a53c9978d55915bda5691be796f\nCloses-Bug: 1826250\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/fd6f5d1926c2536100a9f8394344d0571ac2fc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/fd6f5d1926c2536100a9f8394344d0571ac2fc6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
