)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~658257","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I833e21b1d7bf47121e75ebc07baeb374f5876a42","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/el8-support","hashtags":[],"change_id":"I833e21b1d7bf47121e75ebc07baeb374f5876a42","subject":"[redhat] Update variables to be version agnostic","status":"MERGED","created":"2019-05-10 04:15:42.000000000","updated":"2019-05-15 23:04:05.000000000","submitted":"2019-05-15 23:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658257-1557961445549-c6786d84","meta_rev_id":"3440f98e2af9893e6767b2107ecb8db8c7516d51","_number":658257,"virtual_id_number":658257,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-05-15 23:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-05-10 07:28:58.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 21:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-15 21:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 07:28:58.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-05-15 21:27:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-15 23:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb2e7967803507929e2afbd76f4b25256406f722","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 04:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"798ad2b8a0a9ca1ae6921552851c4efb846c5b5a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 04:25:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f9bb0e784158c6b90adf35c09e8075a788fb3816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 05:55:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/658257/2/check/openstack-ansible-linters/65ee047/ : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/658257/2/check/openstack-ansible-deploy-aio_metal-debian-stable/9f811aa/ : SUCCESS in 57m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/658257/2/check/openstack-ansible-deploy-aio_metal-centos-7/a124b0c/ : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/57/658257/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/3cc7f14/ : SUCCESS in 1h 07m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/658257/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5d4ce8e/ : SUCCESS in 56m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/658257/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4981e23/ : FAILURE in 39m 49s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/57/658257/2/check/openstack-tox-docs/2752a84/html/ : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2af459ed64e13ab066ebb7d1446fe658ddf570f5","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-05-10 07:28:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9a4395cc7cbcf06ce24722288653d2c1c9239494","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-15 21:27:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c6e8e204f5e0304e8a7447795835914500c9376","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 21:28:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a8375744a68ddacaeeef5e4aa958d00bd09b0f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:04:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/658257/2/gate/openstack-ansible-linters/5267d97/ : SUCCESS in 5m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/658257/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/f20982f/ : SUCCESS in 57m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/658257/2/gate/openstack-ansible-deploy-aio_metal-centos-7/01283a9/ : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/57/658257/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/ba58241/ : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/658257/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c6e34f0/ : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/658257/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/903ebe2/ : SUCCESS in 56m 07s\n- openstack-tox-docs http://logs.openstack.org/57/658257/2/gate/openstack-tox-docs/2901666/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f62986377cbf1519aca92a114710ea0f16fe0702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 23:04:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9aa4c5a30f54e23665cad9c3396c34d3b8a3287","revisions":{"97f2c59c029ac2b5d8b5b3aae5f868ab4243272f":{"kind":"REWORK","_number":1,"created":"2019-05-10 04:15:42.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/57/658257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/57/658257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/1"}}},"commit":{"parents":[{"commit":"aa5acfa015478f760111d1df1fa63b1cdb6ffee8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/aa5acfa015478f760111d1df1fa63b1cdb6ffee8"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:15:32.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:15:32.000000000","tz":-240},"subject":"[redhat] Update variables to be version agnostic","message":"[redhat] Update variables to be version agnostic\n\nThe RedHat-based operating systems such as CentOS have a pretty\nstable list of packages, therefore, we don\u0027t need to pin it by\nversion and we can instead move to a much more generic redhat.yml\nwhich will support a bigger range of systems.\n\nChange-Id: I833e21b1d7bf47121e75ebc07baeb374f5876a42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/97f2c59c029ac2b5d8b5b3aae5f868ab4243272f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/97f2c59c029ac2b5d8b5b3aae5f868ab4243272f"}]},"branch":"refs/heads/master"},"f9aa4c5a30f54e23665cad9c3396c34d3b8a3287":{"kind":"REWORK","_number":2,"created":"2019-05-10 04:25:15.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/57/658257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/57/658257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/57/658257/2"}}},"commit":{"parents":[{"commit":"aa5acfa015478f760111d1df1fa63b1cdb6ffee8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/aa5acfa015478f760111d1df1fa63b1cdb6ffee8"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:15:32.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-10 04:25:10.000000000","tz":-240},"subject":"[redhat] Update variables to be version agnostic","message":"[redhat] Update variables to be version agnostic\n\nThe RedHat-based operating systems such as CentOS have a pretty\nstable list of packages, therefore, we don\u0027t need to pin it by\nversion and we can instead move to a much more generic redhat.yml\nwhich will support a bigger range of systems.\n\nChange-Id: I833e21b1d7bf47121e75ebc07baeb374f5876a42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f9aa4c5a30f54e23665cad9c3396c34d3b8a3287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f9aa4c5a30f54e23665cad9c3396c34d3b8a3287"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
