)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~663829","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I27b4ad5cff1752df9b651b2ab6932ee190ec2dee","project":"openstack/openstack-ansible-os_cinder","branch":"master","hashtags":[],"change_id":"I27b4ad5cff1752df9b651b2ab6932ee190ec2dee","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-06-07 02:51:01.000000000","updated":"2019-07-10 18:49:31.000000000","submitted":"2019-07-10 18:49:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663829-1562784571864-6e4dd5bd","meta_rev_id":"e357488e6ea75a392e2dbf3388b1c2597e073806","_number":663829,"virtual_id_number":663829,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 18:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-10 18:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-10 18:37:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 18:37:10.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 18:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0a03ba6ec14e0cb659116fe547c5bc48e6a27c6","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-06-07 02:51:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a47fa3ed42a06ba10e080cb6c326eee6d5ef24b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 03:02:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/29/663829/1/check/requirements-check/b82b605/ : SUCCESS in 1m 58s\n- openstack-ansible-linters http://logs.openstack.org/29/663829/1/check/openstack-ansible-linters/f3d9ad5/ : SUCCESS in 4m 48s\n- openstack-tox-docs http://logs.openstack.org/29/663829/1/check/openstack-tox-docs/42049d5/html/ : SUCCESS in 3m 03s\n- build-openstack-releasenotes http://logs.openstack.org/29/663829/1/check/build-openstack-releasenotes/cc32b01/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a62b6c03be37939e82b0131b718a36900ba71395","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 18:37:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e7ff894e61dde91f6d1f1ad6dec91bc7a99cffb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:37:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"521503f720feb8d9a625fc4f20015f11a63ec9fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:49:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e357488e6ea75a392e2dbf3388b1c2597e073806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:49:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/29/663829/1/gate/requirements-check/a2daaed/ : SUCCESS in 3m 52s\n- openstack-ansible-linters http://logs.openstack.org/29/663829/1/gate/openstack-ansible-linters/28769aa/ : SUCCESS in 9m 40s\n- openstack-tox-docs http://logs.openstack.org/29/663829/1/gate/openstack-tox-docs/511f5df/html/ : SUCCESS in 3m 12s\n- build-openstack-releasenotes http://logs.openstack.org/29/663829/1/gate/build-openstack-releasenotes/4939332/html/ : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16ee70327b5309e93da354db28d3842985f98488","revisions":{"16ee70327b5309e93da354db28d3842985f98488":{"kind":"REWORK","_number":1,"created":"2019-06-07 02:51:01.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/29/663829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/29/663829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/29/663829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/29/663829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/29/663829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/29/663829/1"}}},"commit":{"parents":[{"commit":"0f4b8f276007d287c46a3c032796710cf002f6a9","subject":"db_setup: refactor database setup to a common file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0f4b8f276007d287c46a3c032796710cf002f6a9"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-06-07 02:50:06.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-06-07 02:50:06.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: I27b4ad5cff1752df9b651b2ab6932ee190ec2dee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/16ee70327b5309e93da354db28d3842985f98488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/16ee70327b5309e93da354db28d3842985f98488"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
