)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~664084","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fqueens~I83c1b102afc6a0ca2b2c170f7eb9b34f79fa83db","project":"openstack/openstack-ansible-os_cinder","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I83c1b102afc6a0ca2b2c170f7eb9b34f79fa83db","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:23:23.000000000","updated":"2019-06-11 22:03:59.000000000","submitted":"2019-06-11 22:03:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664084-1560290639873-b5d80b2d","meta_rev_id":"74a861783f0b171fc80375bbd4cb595bc957d176","_number":664084,"virtual_id_number":664084,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 22:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 20:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 20:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:56:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:03:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0c32784c6e4db9e80fd853e745b072426704ee5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a39c1ff300ecb7e38b243a2bf3fee02c117a53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:35:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/84/664084/1/check/build-openstack-sphinx-docs/a5ac485/html/ : SUCCESS in 4m 02s\n- openstack-ansible-linters http://logs.openstack.org/84/664084/1/check/openstack-ansible-linters/a124b5c/ : SUCCESS in 5m 57s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/664084/1/check/openstack-ansible-functional-centos-7/398ec49/ : SUCCESS in 57m 48s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/84/664084/1/check/openstack-ansible-functional-opensuse-423/cc46535/ : FAILURE in 9m 42s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/84/664084/1/check/openstack-ansible-functional-ubuntu-xenial/0654386/ : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2ce48f8e4566688edb53741e8574463734b58159","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 19:31:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0d1210be95e5df0ee529a383accae887735426d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/84/664084/1/check/build-openstack-sphinx-docs/c780d8a/html/ : SUCCESS in 3m 38s\n- openstack-ansible-linters http://logs.openstack.org/84/664084/1/check/openstack-ansible-linters/05b8b32/ : SUCCESS in 5m 41s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/664084/1/check/openstack-ansible-functional-centos-7/15bc2aa/ : SUCCESS in 57m 39s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/84/664084/1/check/openstack-ansible-functional-ubuntu-xenial/61ac57e/ : SUCCESS in 33m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3fb5c6b78506f026882054a7e6ead5dd5f2278dc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:56:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f065476ca1616bbf893002b61b577adbb6281c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:56:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50c2c8fb0647d1c3943f37fa8c6cf7323e89d535","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:03:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/84/664084/1/gate/build-openstack-sphinx-docs/5e8b090/html/ : SUCCESS in 3m 35s\n- openstack-ansible-linters http://logs.openstack.org/84/664084/1/gate/openstack-ansible-linters/afc9755/ : SUCCESS in 5m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/84/664084/1/gate/openstack-ansible-functional-centos-7/e6b7dfd/ : SUCCESS in 1h 01m 22s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/84/664084/1/gate/openstack-ansible-functional-ubuntu-xenial/f2bb10b/ : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"785fd689edefe2e3e706e65c194c65d36dab2009","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:03:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b180454346cc84dc2e70a14d95ba2ea5614928e8","revisions":{"b180454346cc84dc2e70a14d95ba2ea5614928e8":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:23:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/664084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/84/664084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/84/664084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/84/664084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/84/664084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/84/664084/1"}}},"commit":{"parents":[{"commit":"4fe4963517cde2fa3bda54bc3fa3320e786355c4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/4fe4963517cde2fa3bda54bc3fa3320e786355c4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:23:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:23:19.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I83c1b102afc6a0ca2b2c170f7eb9b34f79fa83db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b180454346cc84dc2e70a14d95ba2ea5614928e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b180454346cc84dc2e70a14d95ba2ea5614928e8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
