)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~664309","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Frocky~Ife5ca4521028f00f4bf9d923dfd5d0196a15205e","project":"openstack/openstack-ansible-os_cinder","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ife5ca4521028f00f4bf9d923dfd5d0196a15205e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:44:02.000000000","updated":"2019-06-10 18:18:13.000000000","submitted":"2019-06-10 18:18:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664309-1560190693883-2d20edbd","meta_rev_id":"cce6de439e70e10fbbd25198bad6788d69178a8c","_number":664309,"virtual_id_number":664309,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 18:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:16:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:16:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 18:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cc5fb3af037201069fd7bcf61dfc24fa66b820a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55f8097125c0933da2a4d8ff8b40f32009775872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:45:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/664309/1/check/openstack-ansible-linters/d77a283/ : SUCCESS in 5m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-centos-7/c9fb4eb/ : SUCCESS in 47m 58s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-opensuse-423/57f36fb/ : FAILURE in 8m 20s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-opensuse-150/b57165a/ : FAILURE in 10m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-ubuntu-bionic/2e680e2/ : SUCCESS in 33m 21s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-ubuntu-xenial/24ed68d/ : SUCCESS in 33m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/664309/1/check/build-openstack-sphinx-docs/b8626cc/html/ : SUCCESS in 3m 35s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-distro_install-centos-7/d55b116/ : SUCCESS in 33m 09s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-distro_install-opensuse-423/4493327/ : FAILURE in 8m 19s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/09/664309/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/1b06266/ : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"400ebfa4a526fd3a300c8fb8b239504744dfc766","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:16:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"651eeb94b6e065ef5f300a898990089dd7e5a049","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:16:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36ab944a487f074ed2ac0f1b76f14c9a98c3948c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:18:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4834406e36c7c38c6aeb7f1ff1521d39c2690c66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:18:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/664309/1/gate/openstack-ansible-linters/b95f3d8/ : SUCCESS in 4m 16s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-centos-7/396d5e8/ : SUCCESS in 45m 28s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-opensuse-423/7712dc1/ : FAILURE in 8m 32s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-ubuntu-bionic/47984d5/ : SUCCESS in 38m 36s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-ubuntu-xenial/5955271/ : SUCCESS in 40m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/09/664309/1/gate/build-openstack-sphinx-docs/2919a94/html/ : SUCCESS in 4m 00s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-distro_install-centos-7/8301e49/ : SUCCESS in 41m 44s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-distro_install-opensuse-423/2ed4dc1/ : FAILURE in 9m 03s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/09/664309/1/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/1d76325/ : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2f327ae21d36efc884dcf0136f14d28e21106b2","revisions":{"a2f327ae21d36efc884dcf0136f14d28e21106b2":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/664309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/09/664309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/664309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/664309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/664309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/664309/1"}}},"commit":{"parents":[{"commit":"0e3c7be52cb23764e025803f0d03ba5cc2bfe12b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/0e3c7be52cb23764e025803f0d03ba5cc2bfe12b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:01.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ife5ca4521028f00f4bf9d923dfd5d0196a15205e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a2f327ae21d36efc884dcf0136f14d28e21106b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a2f327ae21d36efc884dcf0136f14d28e21106b2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
