)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~664717","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I1b5ccc351116c2530013b26ce7a8194c61344ed9","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I1b5ccc351116c2530013b26ce7a8194c61344ed9","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:17:21.000000000","updated":"2019-06-12 07:54:20.000000000","submitted":"2019-06-12 07:54:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664717-1560326060951-8e310be5","meta_rev_id":"3bdfbaa4b08e5b2c720ca3dd7bbfce1d9dc27992","_number":664717,"virtual_id_number":664717,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-12 07:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 06:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 06:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:15:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 07:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a39ccf97cc129f0181c7dfcc5b01e0981ce29405","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb5fa9804b0bbe49049a3592723937e7abc93309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/17/664717/1/check/openstack-ansible-linters/4724d9a/ : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/17/664717/1/check/openstack-ansible-deploy-aio_metal-debian-stable/15a157b/ : SUCCESS in 1h 03m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/17/664717/1/check/openstack-ansible-deploy-aio_metal-centos-7/178c41e/ : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/17/664717/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/543d6db/ : SUCCESS in 1h 03m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/17/664717/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0cdd251/ : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/17/664717/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8ac3634/ : SUCCESS in 54m 50s\n- openstack-tox-docs http://logs.openstack.org/17/664717/1/check/openstack-tox-docs/9eb631f/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f49362cd049aeaed87353a9d38dfab7ceb7daca8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 06:15:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5431fa4f9e32e5b6bac832342331f9938be1358e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:15:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d432cbe0f909367c600528bf1bd5629d5d43d5f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:54:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1745990a67c2cfb60f3df9ec6026817b148b502a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:54:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/17/664717/1/gate/openstack-ansible-linters/ee2b99a/ : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/17/664717/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/8b025a3/ : SUCCESS in 59m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/17/664717/1/gate/openstack-ansible-deploy-aio_metal-centos-7/b498ac1/ : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/17/664717/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/12e6f40/ : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/17/664717/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/71aff1d/ : SUCCESS in 1h 09m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/17/664717/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/eb2564c/ : SUCCESS in 1h 00m 19s\n- openstack-tox-docs http://logs.openstack.org/17/664717/1/gate/openstack-tox-docs/c1e0f73/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69a77bce05f7a6cb86502c2dc13949444fc03801","revisions":{"69a77bce05f7a6cb86502c2dc13949444fc03801":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:17:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/664717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/17/664717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/17/664717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/17/664717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/17/664717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/17/664717/1"}}},"commit":{"parents":[{"commit":"d139d40a19307230bd5e3dc2580919fe9f6f16c1","subject":"Cleanup debian cinder service distro package list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/d139d40a19307230bd5e3dc2580919fe9f6f16c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:17:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:17:19.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I1b5ccc351116c2530013b26ce7a8194c61344ed9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/69a77bce05f7a6cb86502c2dc13949444fc03801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/69a77bce05f7a6cb86502c2dc13949444fc03801"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
