)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~665612","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I4755fb3cfed4707cd06299c2ff97a5011c62f147","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"static_imports","hashtags":[],"change_id":"I4755fb3cfed4707cd06299c2ff97a5011c62f147","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-06-17 08:48:43.000000000","updated":"2019-06-28 13:49:18.000000000","submitted":"2019-06-28 13:49:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665612-1561729758702-a45a0009","meta_rev_id":"58366c9931b2624eb019d0b105dcb7388280642d","_number":665612,"virtual_id_number":665612,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-28 13:49:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":2,"date":"2019-06-28 10:26:45.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-28 12:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-28 12:17:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 10:26:45.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-28 12:17:01.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-28 13:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0aca3a613ef709f20011fef5f0ba91fd17e613e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-17 08:48:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e7ea3b5ba816600d18cd182bb65ceb6c324c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 10:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/12/665612/1/check/openstack-ansible-linters/db0d3e1/ : SUCCESS in 4m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/12/665612/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1f09f3a/ : SUCCESS in 59m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/12/665612/1/check/openstack-ansible-deploy-aio_metal-centos-7/220368b/ : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/12/665612/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/ddb6ad0/ : SUCCESS in 1h 05m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/12/665612/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ffd47fa/ : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/12/665612/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/79a3a61/ : SUCCESS in 1h 00m 19s\n- openstack-tox-docs http://logs.openstack.org/12/665612/1/check/openstack-tox-docs/9290b46/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"00efbbc374652b4cc1e1b854e3850620cc5c7fcc","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-28 10:26:45.000000000","message":"Patch Set 1: Code-Review+2\n\nWe try this on a regular basis, but I think we should do it, and deal with the pain.","accounts_in_message":[],"_revision_number":1},{"id":"8270b33413a3d485d04211d23089c024e2ad19ec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 12:17:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d96a1791c8de2245e3d6048b89e95a340851f878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 12:17:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3e5f01efd2b7b391f752f4e4c14640e13927146","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:49:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b11672167b34de19d273198c1a49805669d64183","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 13:49:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/12/665612/1/gate/openstack-ansible-linters/931080a/ : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/12/665612/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/e7563e3/ : SUCCESS in 1h 04m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/12/665612/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6c5c4a8/ : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/12/665612/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/e2a0d14/ : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/12/665612/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/525f85b/ : SUCCESS in 56m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/12/665612/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/2aa55c4/ : SUCCESS in 59m 04s\n- openstack-tox-docs http://logs.openstack.org/12/665612/1/gate/openstack-tox-docs/f601e7f/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb98ba104e5b28ae6a4a3c3ef2ea5053fc71ce6d","revisions":{"bb98ba104e5b28ae6a4a3c3ef2ea5053fc71ce6d":{"kind":"REWORK","_number":1,"created":"2019-06-17 08:48:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/665612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/12/665612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/12/665612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/12/665612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/12/665612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/12/665612/1"}}},"commit":{"parents":[{"commit":"69a77bce05f7a6cb86502c2dc13949444fc03801","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/69a77bce05f7a6cb86502c2dc13949444fc03801"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-17 08:48:29.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-06-17 08:48:29.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like neutron-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I4755fb3cfed4707cd06299c2ff97a5011c62f147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/bb98ba104e5b28ae6a4a3c3ef2ea5053fc71ce6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/bb98ba104e5b28ae6a4a3c3ef2ea5053fc71ce6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
