)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~666677","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Frocky~I6899fe983dc21c2090a6755dacb5ee174814611f","project":"openstack/openstack-ansible-os_cinder","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6899fe983dc21c2090a6755dacb5ee174814611f","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:00:14.000000000","updated":"2019-06-20 22:27:56.000000000","submitted":"2019-06-20 22:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666677-1561069676153-2d1303b5","meta_rev_id":"c3f2f07beb9dd82a018780fd6d5e9b285cfa6d2a","_number":666677,"virtual_id_number":666677,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-20 22:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 21:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 21:25:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:07:07.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-20 21:25:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-20 22:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8522e9a34b2211f70b635c4fc3a7cfb4c55323dd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d09051d191379557ebfd6c8cf611979a44bd720","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:07:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4604521a415f06f493e2466020cfbcda59c94904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:07:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/666677/1/check/openstack-ansible-linters/170c899/ : SUCCESS in 5m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-centos-7/b87c46b/ : SUCCESS in 44m 23s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-opensuse-423/0c57787/ : FAILURE in 6m 36s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-opensuse-150/db3816a/ : FAILURE in 7m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-ubuntu-bionic/b9a3d26/ : SUCCESS in 1h 02m 26s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-ubuntu-xenial/b9fcce4/ : SUCCESS in 35m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/77/666677/1/check/build-openstack-sphinx-docs/17c995a/html/ : SUCCESS in 3m 15s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-distro_install-centos-7/8f18447/ : SUCCESS in 39m 56s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-distro_install-opensuse-423/518f79f/ : FAILURE in 8m 56s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/77/666677/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/2c85134/ : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5cf25b04b9cec325b71b4015e8ba019d4434ca2e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 21:25:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06159787239ad7200f59c00ba2e54414e9a3579d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:25:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"409a9791103d70f5f4416ea493eea1d2d66d14d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:27:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/666677/1/gate/openstack-ansible-linters/5019b91/ : SUCCESS in 5m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-centos-7/262b35e/ : SUCCESS in 42m 57s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-opensuse-423/e794e40/ : FAILURE in 6m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-ubuntu-bionic/af95769/ : SUCCESS in 41m 01s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-ubuntu-xenial/8b280db/ : SUCCESS in 35m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/77/666677/1/gate/build-openstack-sphinx-docs/e8824f3/html/ : SUCCESS in 4m 23s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-distro_install-centos-7/246a4b2/ : SUCCESS in 42m 15s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-distro_install-opensuse-423/068083d/ : FAILURE in 6m 02s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/77/666677/1/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/6729ee6/ : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c704cac4f6259ed4b97e26dc6004bed5c188c99e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:27:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61b32fc84057f2340f68cbc4d37d3a7b0cdff3ba","revisions":{"61b32fc84057f2340f68cbc4d37d3a7b0cdff3ba":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:00:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/666677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/77/666677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/77/666677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/77/666677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/77/666677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/77/666677/1"}}},"commit":{"parents":[{"commit":"a2f327ae21d36efc884dcf0136f14d28e21106b2","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/a2f327ae21d36efc884dcf0136f14d28e21106b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:00:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:00:12.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6899fe983dc21c2090a6755dacb5ee174814611f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/61b32fc84057f2340f68cbc4d37d3a7b0cdff3ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/61b32fc84057f2340f68cbc4d37d3a7b0cdff3ba"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
