)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~666700","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fstein~I6e537e43353740afe1f24682694f40901530ea56","project":"openstack/openstack-ansible-os_cinder","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6e537e43353740afe1f24682694f40901530ea56","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-20 20:04:02.000000000","updated":"2019-06-20 23:37:14.000000000","submitted":"2019-06-20 23:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666700-1561073834636-eb0806c3","meta_rev_id":"1d88c7cdb75c92c3d17872f602a0690ed72eba50","_number":666700,"virtual_id_number":666700,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-20 23:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 22:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 22:10:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:08:36.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-20 22:10:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-20 23:37:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa4886af0ce83fe24325759f89aff666d88fa08f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e8e0e1893814e2fc6f6192f3264bf264ad00355","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:08:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16cd961a6e920164029cbe9a5823f053c719f74c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:44:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/666700/1/check/openstack-ansible-linters/c25de3a/ : SUCCESS in 8m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/00/666700/1/check/openstack-ansible-deploy-aio_metal-debian-stable/22f910f/ : SUCCESS in 56m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/00/666700/1/check/openstack-ansible-deploy-aio_metal-centos-7/2788796/ : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/00/666700/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c1b0f17/ : SUCCESS in 57m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/00/666700/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/93a4ddb/ : FAILURE in 38m 10s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/00/666700/1/check/openstack-tox-docs/726e7a5/html/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"af784b5db7ffdb069b034dbc1df00ffa0440a9fb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 22:10:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8df916fdc05d66ff535e648cbe80e6588f7683db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:11:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4028a4cefe43bd55c68b1514e316381dc7f3494d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:37:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7118afe2e453d412969340da42763e5e00dab19f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/666700/1/gate/openstack-ansible-linters/85c5d04/ : SUCCESS in 5m 44s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/00/666700/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/d15e486/ : SUCCESS in 56m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/00/666700/1/gate/openstack-ansible-deploy-aio_metal-centos-7/3e929ae/ : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/00/666700/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7be8d70/ : SUCCESS in 1h 02m 37s\n- openstack-tox-docs http://logs.openstack.org/00/666700/1/gate/openstack-tox-docs/103f86c/html/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eefb199ae4d70d12f5b79224cd66527bdad77799","revisions":{"eefb199ae4d70d12f5b79224cd66527bdad77799":{"kind":"REWORK","_number":1,"created":"2019-06-20 20:04:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/666700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/00/666700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/666700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/666700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/666700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/00/666700/1"}}},"commit":{"parents":[{"commit":"2b274c775806c1cbe015a718afd0ee6f9f0b5960","subject":"release: added release note for ceph backend issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/2b274c775806c1cbe015a718afd0ee6f9f0b5960"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:04:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:04:00.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6e537e43353740afe1f24682694f40901530ea56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/eefb199ae4d70d12f5b79224cd66527bdad77799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/eefb199ae4d70d12f5b79224cd66527bdad77799"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
