)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~671811","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I3bd2cd503d6dd12edd23026a2d588e56aab39ac1","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I3bd2cd503d6dd12edd23026a2d588e56aab39ac1","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-19 18:20:16.000000000","updated":"2019-07-20 17:34:29.000000000","submitted":"2019-07-20 17:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671811-1563644069877-6f1d36b5","meta_rev_id":"30417ee101f284b79386f5657439d5835e1518a2","_number":671811,"virtual_id_number":671811,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-20 17:34:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-20 17:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-20 17:21:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-20 17:21:13.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-20 17:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8742449d36b690d1da8392dce25228781fcc1eca","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dbf544c1768f613d2d6b44adbcec1d13f3c8917","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:41:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/11/671811/1/check/requirements-check/d6f5df9/ : SUCCESS in 2m 47s\n- openstack-ansible-linters http://logs.openstack.org/11/671811/1/check/openstack-ansible-linters/109baff/ : SUCCESS in 6m 47s\n- openstack-tox-docs http://logs.openstack.org/11/671811/1/check/openstack-tox-docs/2dd032c/ : POST_FAILURE in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/11/671811/1/check/build-openstack-releasenotes/de1929c/html/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"af3e8206aacf8f13ddded5df6b41a4e82e6600e2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:30:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dee9560d574a2ab95dd004e65578867a9c44c80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:42:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/11/671811/2/check/requirements-check/5b1eefc/ : SUCCESS in 2m 18s\n- openstack-ansible-linters http://logs.openstack.org/11/671811/2/check/openstack-ansible-linters/76bf147/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/11/671811/2/check/openstack-tox-docs/f52fb27/html/ : SUCCESS in 3m 03s\n- build-openstack-releasenotes http://logs.openstack.org/11/671811/2/check/build-openstack-releasenotes/9ab3529/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7137ff851cbfa5e5eb7a7fc9cf384039eead15a9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-20 17:21:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf81c61b1849b9e26cee6b77c5df1522dd6700ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 17:21:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4d1fa6911b5463901ebf22e2a5648d7df97092f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 17:34:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"12ce9f10fdd26e34eee5348b1e5c6dbf6cbe94a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 17:34:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/11/671811/2/gate/requirements-check/5fd776e/ : SUCCESS in 2m 59s\n- openstack-ansible-linters http://logs.openstack.org/11/671811/2/gate/openstack-ansible-linters/459fb3d/ : SUCCESS in 8m 41s\n- openstack-tox-docs http://logs.openstack.org/11/671811/2/gate/openstack-tox-docs/d5f1e94/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes http://logs.openstack.org/11/671811/2/gate/build-openstack-releasenotes/98bb561/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c9bae0821dc76fa10a2f213d94cceffbc8764d96","revisions":{"30f10ec08e8797fe00bcc67addf789126c61d4da":{"kind":"REWORK","_number":1,"created":"2019-07-19 18:20:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/671811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/671811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/1"}}},"commit":{"parents":[{"commit":"f623641f972000f96f945579a3975eb1a334f19e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f623641f972000f96f945579a3975eb1a334f19e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:20:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:20:11.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3bd2cd503d6dd12edd23026a2d588e56aab39ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/30f10ec08e8797fe00bcc67addf789126c61d4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/30f10ec08e8797fe00bcc67addf789126c61d4da"}]},"branch":"refs/heads/master"},"c9bae0821dc76fa10a2f213d94cceffbc8764d96":{"kind":"REWORK","_number":2,"created":"2019-07-20 09:30:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/671811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/11/671811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/11/671811/2"}}},"commit":{"parents":[{"commit":"f623641f972000f96f945579a3975eb1a334f19e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/f623641f972000f96f945579a3975eb1a334f19e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:30:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:30:51.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I3bd2cd503d6dd12edd23026a2d588e56aab39ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c9bae0821dc76fa10a2f213d94cceffbc8764d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c9bae0821dc76fa10a2f213d94cceffbc8764d96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
