)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~673275","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~I6ece609161f6252ce5a0ddb48d1e66247049d675","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa/service-refactor","hashtags":[],"change_id":"I6ece609161f6252ce5a0ddb48d1e66247049d675","subject":"service_setup: Update services tasks","status":"MERGED","created":"2019-07-29 11:57:38.000000000","updated":"2019-07-29 22:31:32.000000000","submitted":"2019-07-29 22:31:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673275-1564439492257-8da850ff","meta_rev_id":"7ef9c309b68aedf1e1bd5b9729ab785a39dd1e35","_number":673275,"virtual_id_number":673275,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-29 22:31:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-29 20:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-29 20:50:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-29 20:50:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 20:47:33.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-29 20:50:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-29 22:31:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6628142117fbb955c938ddb4ed3cf5cbd163e856","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-29 11:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9213035109ea1003512dc8588335ebed1e69378","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 13:35:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/75/673275/1/check/openstack-ansible-linters/75d5c34/ : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7d3c08b/ : SUCCESS in 53m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_metal-centos-7/df13d1c/ : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/5586b81/ : SUCCESS in 1h 14m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0a8ca6f/ : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3a07e5b/ : SUCCESS in 1h 13m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/039e682/ : FAILURE in 46m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/75/673275/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/20145dd/ : SUCCESS in 39m 13s\n- openstack-tox-docs http://logs.openstack.org/75/673275/1/check/openstack-tox-docs/4140e37/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"68b3dc48b776ffca9e37fdb7951395c3e8d143f7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-29 20:47:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38f2f47ab4dca857e9cd41e20ebf39ab870e8dd7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-29 20:50:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d139b07c00a086e256a9b07923b31e8bf00a9941","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:50:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebd5ed0f35e799cc02057a1e83276fd5534dfe12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:31:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f519319f8f2bcfa01ba1b418464a33eca79c563","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:31:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/75/673275/1/gate/openstack-ansible-linters/7f42fe4/ : SUCCESS in 8m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/05d6ea6/ : SUCCESS in 1h 02m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_metal-centos-7/def4eac/ : SUCCESS in 1h 31m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/edcd576/ : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dbc6ed1/ : SUCCESS in 57m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/755868b/ : SUCCESS in 59m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/75/673275/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/23c1c5b/ : SUCCESS in 43m 50s\n- openstack-tox-docs http://logs.openstack.org/75/673275/1/gate/openstack-tox-docs/a15d6a8/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95c76200093a6d8d6ae653723049ea21f1192d08","revisions":{"95c76200093a6d8d6ae653723049ea21f1192d08":{"kind":"REWORK","_number":1,"created":"2019-07-29 11:57:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/673275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/75/673275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/75/673275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/75/673275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/75/673275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/75/673275/1"}}},"commit":{"parents":[{"commit":"c148d77e29af6faebc1c9b012ae08aed447cd179","subject":"rbd: start using active-active","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/c148d77e29af6faebc1c9b012ae08aed447cd179"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-29 11:57:27.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-29 11:57:27.000000000","tz":180},"subject":"service_setup: Update services tasks","message":"service_setup: Update services tasks\n\nDuring distribution of service_setup.yml it appeared that some services\nhave extra requirments and usecases. So this patch updated service_setup\nand brings it into accordance across the roles.\n\nChange-Id: I6ece609161f6252ce5a0ddb48d1e66247049d675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/95c76200093a6d8d6ae653723049ea21f1192d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/95c76200093a6d8d6ae653723049ea21f1192d08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
