)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~674610","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fstein~I0c8aed4d0608c1f117e1baa1f428875956159ffd","project":"openstack/openstack-ansible-os_cinder","branch":"stable/stein","topic":"bug/1837403-stable/stein","hashtags":[],"change_id":"I0c8aed4d0608c1f117e1baa1f428875956159ffd","subject":"rbd: start using active-active","status":"MERGED","created":"2019-08-05 13:45:05.000000000","updated":"2019-08-07 19:01:34.000000000","submitted":"2019-08-07 19:01:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674610-1565204494966-cf54861c","meta_rev_id":"23e4d34b68dd4aa9ff96d2c64add31968786238b","_number":674610,"virtual_id_number":674610,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-07 19:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-08-07 17:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-07 17:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-07 17:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 17:12:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-07 17:32:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-07 19:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffe5278369e1b63e16fdc0c5afe3ea7405b3003c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-05 13:45:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5e11d12eab07e94b8fdb9bd3328dee1c5f0ee431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:42:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/10/674610/1/check/openstack-ansible-linters/dd3ae1d/ : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/10/674610/1/check/openstack-ansible-deploy-aio_metal-debian-stable/e48e468/ : SUCCESS in 1h 01m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/10/674610/1/check/openstack-ansible-deploy-aio_metal-centos-7/e554c86/ : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/10/674610/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8ead171/ : SUCCESS in 59m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/10/674610/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ead469f/ : SUCCESS in 59m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/10/674610/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/35a10f9/ : SUCCESS in 41m 58s\n- openstack-tox-docs https://logs.opendev.org/10/674610/1/check/openstack-tox-docs/2d2ae4c/html/ : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://logs.opendev.org/10/674610/1/check/build-openstack-releasenotes/bd04775/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d2dfbec89ffc8935bb567d55f1697fb94d6cc3ae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 17:12:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f182dad47b6198ae9dc2ae60806ed153bcba502","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-08-07 17:32:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5837bd779c9178eca92b56cf2e5eaf0daaa8878e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:35:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"feeaf61ec3c21a5ffc4f4663e47eb3804fcf13de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:01:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee4040826045b1bbf5c76e51c90bbe22f53d4120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:01:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/10/674610/1/gate/openstack-ansible-linters/84ace3c/ : SUCCESS in 5m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/10/674610/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/4e72901/ : SUCCESS in 1h 04m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/10/674610/1/gate/openstack-ansible-deploy-aio_metal-centos-7/aba3d18/ : SUCCESS in 1h 23m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/10/674610/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/4619111/ : SUCCESS in 1h 02m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/10/674610/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/7bb47c0/ : SUCCESS in 59m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/10/674610/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8f97459/ : SUCCESS in 47m 23s\n- openstack-tox-docs https://logs.opendev.org/10/674610/1/gate/openstack-tox-docs/aaa87d6/html/ : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://logs.opendev.org/10/674610/1/gate/build-openstack-releasenotes/4daab5e/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"918b9077c816be5fc056637301265e0be2f245ab","revisions":{"918b9077c816be5fc056637301265e0be2f245ab":{"kind":"REWORK","_number":1,"created":"2019-08-05 13:45:05.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/10/674610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/10/674610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/10/674610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/10/674610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/10/674610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/10/674610/1"}}},"commit":{"parents":[{"commit":"6f03ab540fa6bc558a410d3fbbc2f3f6e1d1b15a","subject":"Convert dynamic includes to static imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/6f03ab540fa6bc558a410d3fbbc2f3f6e1d1b15a"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-22 16:10:01.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-08-05 13:45:05.000000000","tz":0},"subject":"rbd: start using active-active","message":"rbd: start using active-active\n\nThis patch drops the hacky workaround of using backend_host which\nis not recommended by the Cinder team and instead uses active-active\nRBD which has been implemented since Rocky.\n\nCloses-Bug: #1837403\nChange-Id: I0c8aed4d0608c1f117e1baa1f428875956159ffd\n(cherry picked from commit c148d77e29af6faebc1c9b012ae08aed447cd179)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/918b9077c816be5fc056637301265e0be2f245ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/918b9077c816be5fc056637301265e0be2f245ab"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
