)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~683195","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~master~Ib51d7c1a119f6d2890d94d1fcb3e7d4b8d76fc91","project":"openstack/openstack-ansible-os_cinder","branch":"master","topic":"osa_leap15_python3","hashtags":[],"change_id":"Ib51d7c1a119f6d2890d94d1fcb3e7d4b8d76fc91","subject":"Fix distro install on openSUSE","status":"MERGED","created":"2019-09-19 16:00:22.000000000","updated":"2019-09-26 16:53:32.000000000","submitted":"2019-09-26 16:52:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683195-1569516741033-b42f53c0","meta_rev_id":"dae3197676b9c311f867a72a95ed88bdaca4b1d6","_number":683195,"virtual_id_number":683195,"owner":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 16:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 13:14:28.000000000","_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-20 13:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":2,"date":"2019-09-20 14:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":1,"date":"2019-09-20 14:40:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 13:44:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-20 14:40:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-26 16:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b92a1f0abe235e988fac8d18cbeeef6c60f3105","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-19 16:00:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ff4dad584c3b31c9068aec19a4a04552df099a5","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-19 16:01:04.000000000","message":"Topic set to osa_leap15_python3","accounts_in_message":[],"_revision_number":1},{"id":"005ba172140eb210e2a90aa2a996054d41d41b0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:57:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7b762d35c904bcc9e416c6d4d7ffd91 : SUCCESS in 7m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/25ef4347950d4dcc84203a3a145ca5d0 : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/80fd7c5d527744d99f10007c57f611c7 : SUCCESS in 1h 48m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8b81067211d7422aae76e62a8d142112 : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53473db749ba456599b9596231c65bb7 : SUCCESS in 1h 13m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c43d7ae4f434539a02b9f1e4cbfa48e : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2027da5bcdc9434092c40e0b33378c49 : FAILURE in 31m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53507fc33ee44a058c722a7894e2017c : SUCCESS in 55m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a640cf5b5b4289bb75acd37d91691e : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ffcf8667c0e5cc6497259726b2c8acb3b0a12827","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 13:44:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11103575134a5f9d4f30cfc44429f8cb283a6258","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-20 14:40:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f58af3db4c11c46f003d07fa626c6d50bae583b","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-24 14:23:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83bb0730aef076a8e61ae4ee9d4e6c02440e57be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 16:25:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b749979c22484a478d0cfc63deca69ee : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/efe73fff5f394505a90e3cd528a0ec87 : FAILURE in 42m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ee7cfdfbff02421a9840279079ae7cc9 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/26dfef5055834b69a2e020615e46624c : SUCCESS in 1h 51m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df34f53223d64e308a508bbff99f38b5 : FAILURE in 1h 10m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b0d850790e044ff2a33480a2c4876f82 : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ae28b38bf167424e896fa6c958c8c4f1 : FAILURE in 55m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64dfc78405ff41d781fa9f51b0b3a01a : SUCCESS in 1h 07m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb1760ca24a4b60b7b564f01678b727 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7077d097aa0762cd9933af1d0081627a16412435","author":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"date":"2019-09-26 13:14:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98e061e830f0f8e5ba4e237d085307a17fff25b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 15:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8896683e9eba44869cc6de3a6e75f86c : SUCCESS in 6m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a15d0cd8cb0d4c1185b3c71871ab9358 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3b8a1290a03400f955bfbfd383495dd : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d0de399f051b45bf93ebefa22bfd5f3e : SUCCESS in 1h 38m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203f445a2cd842c3884212ecc5399b8d : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0ca256e549244f18bb9249d51d2a250 : SUCCESS in 1h 20m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c5a18d5c27074bc28171039cdc5efd54 : FAILURE in 1h 03m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc25bec570f54f66a28a24d919f6a462 : SUCCESS in 57m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781558a9601343f8ae77e66fca6311b0 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8fe1841a7e04cd7ffef1bd525eba7d85c1bd2eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 15:08:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffb144067c4f980d0e545e4d07ba2b296c3bf336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:52:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2698ef4e192d40f4856b19d766040819 : SUCCESS in 4m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/042c48b33516452d9d96d9966d7aa1a6 : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b50ea2e8f42548eda7dc797fa6f0a6fb : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2a7fac82558c4ecca04cb70560b318a3 : SUCCESS in 1h 37m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aade6229dc7647608ec824dfd2f9a701 : SUCCESS in 1h 21m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dc9e6fc3c53f436793417783b819109b : SUCCESS in 1h 13m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f50f57d9a97f4ad7b36c23a25d990238 : SUCCESS in 1h 07m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1f3d9f17d14654a88b18663dda90be : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1e02aeca01d9342bf35d076a028eddaaeed4ea04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:52:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dae3197676b9c311f867a72a95ed88bdaca4b1d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 16:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08ab62f2f0a417b8a96e44a27499e35 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da93dc99e828636055cbcee63ae6ba41ab29c93f","revisions":{"da93dc99e828636055cbcee63ae6ba41ab29c93f":{"kind":"REWORK","_number":1,"created":"2019-09-19 16:00:22.000000000","uploader":{"_account_id":2062,"name":"Ralf Haferkamp","email":"rhafer@suse.com","username":"rhafer"},"ref":"refs/changes/95/683195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/95/683195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/683195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/683195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/683195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/95/683195/1"}}},"commit":{"parents":[{"commit":"5681e2d2b50e2c5acaee2fad23591bcf64101544","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/5681e2d2b50e2c5acaee2fad23591bcf64101544"}]}],"author":{"name":"Ralf Haferkamp","email":"rhafer@suse.com","date":"2019-09-19 15:58:40.000000000","tz":120},"committer":{"name":"Ralf Haferkamp","email":"rhafer@suse.com","date":"2019-09-19 16:00:20.000000000","tz":120},"subject":"Fix distro install on openSUSE","message":"Fix distro install on openSUSE\n\nWith Train openSUSE is switching the openstack packages to use Python 3.\nThis means that we need to use the python3 variants of python-module\npackages.\n\nDepends-On: https://review.opendev.org/683123\nChange-Id: Ib51d7c1a119f6d2890d94d1fcb3e7d4b8d76fc91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/da93dc99e828636055cbcee63ae6ba41ab29c93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/da93dc99e828636055cbcee63ae6ba41ab29c93f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
