)]}'
{"id":"openstack%2Fopenstack-ansible-os_cinder~688509","triplet_id":"openstack%2Fopenstack-ansible-os_cinder~stable%2Fpike~I4c08e1336441dc2af6feb86c6c25149635402222","project":"openstack/openstack-ansible-os_cinder","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"I4c08e1336441dc2af6feb86c6c25149635402222","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 17:22:01.000000000","updated":"2019-10-14 20:23:38.000000000","submitted":"2019-10-14 20:23:38.000000000","submitter":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688509-1571084618052-f62c7b6a","meta_rev_id":"e526cacf055ee54d8f0db9c4619dbecca0be92fd","_number":688509,"virtual_id_number":688509,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-14 20:23:34.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-14 18:11:43.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-10-14 17:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-10-14 17:24:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:24:25.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 18:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 20:23:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"b31552630c3866b216733d999a49a6b85d446403","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 17:22:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f52ef0ea6fc4e7ec10f9f8b42fbbf3e08ca8fe88","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:24:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80be5bf969db93bf69688cb73c7895b5770240d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a6b0521867144e64bd293527e47f06bc : SUCCESS in 4m 25s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fe7c890384744a13920fac7db9b10bde : SUCCESS in 7m 30s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/63a37de600b1476498593788f926441d : FAILURE in 8m 05s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0fdc2227f1dd4f6b937e580b5d85b78f : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e23086270c8c7bb38bf60b365e33b04e203d24d7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:23:34.000000000","message":"Patch Set 1: Verified+2\n\nThe opensuse-423 nodeset is being removed.  This change is being force-merged to correct the resulting configuration error.","accounts_in_message":[],"_revision_number":1},{"id":"624a8b3a60ef9da19059509f6336a124bf961a6c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:23:38.000000000","message":"Change has been successfully merged by Clark Boylan","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8c07ccbe929adb4da6125a32a0523c52f70c90b","revisions":{"b8c07ccbe929adb4da6125a32a0523c52f70c90b":{"kind":"REWORK","_number":1,"created":"2019-10-14 17:22:01.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/688509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_cinder","ref":"refs/changes/09/688509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/688509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/688509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/688509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_cinder refs/changes/09/688509/1"}}},"commit":{"parents":[{"commit":"b4acc75816e8481adb5cfc006a5e8b1299e25e76","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b4acc75816e8481adb5cfc006a5e8b1299e25e76"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:21:46.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:21:58.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: I4c08e1336441dc2af6feb86c6c25149635402222\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b8c07ccbe929adb4da6125a32a0523c52f70c90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_cinder/commit/b8c07ccbe929adb4da6125a32a0523c52f70c90b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
